)]}'
{"id":"openstack%2Frally~706896","triplet_id":"openstack%2Frally~master~Ibc901c400e55f201eaaf25eaa1c3eed17b4e0480","project":"openstack/rally","branch":"master","topic":"py3","hashtags":[],"change_id":"Ibc901c400e55f201eaaf25eaa1c3eed17b4e0480","subject":"Fix deprecated things","status":"MERGED","created":"2020-02-10 16:28:22.000000000","updated":"2020-02-15 17:29:16.000000000","submitted":"2020-02-15 17:29:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706896-1581787756549-b4850ae3","meta_rev_id":"0994fe109d0fdf4da6c490c4da3546f97df785b1","_number":706896,"virtual_id_number":706896,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-15 17:29:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-02-15 17:06:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-02-15 17:06:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-15 17:29:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"979168802f8e72c5ef50a89b127c472e1dd5d443","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-10 16:28:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bfbc24f54deedd63abe34700e7d57db3c51425c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 17:01:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/588c70816cb54cdb8e76664ce2029854 : SUCCESS in 5m 50s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/9a9046af64d34edb8d75f868830577c6 : SUCCESS in 7m 27s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2542e5049b094570bd32005ab3499baa : SUCCESS in 5m 10s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/8e82b415ff714b22ab4cac081e6d2081 : SUCCESS in 5m 07s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/ac2c3363a4234ecd8112bff790f7de47 : SUCCESS in 5m 20s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/a0edac3b413944bea32ce5cfee3eea3f : SUCCESS in 6m 03s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/1e475bb645fe43be9279a5a56f5f0e10 : FAILURE in 7m 24s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/cc0dcc33f1e04b31a4f7a6a7ab6795c3 : FAILURE in 9m 26s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/5fcd74ef9c4949eb9f2320f57dd177c3 : FAILURE in 5m 41s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/120871544a3d49b4a1c9f883cff042b7 : FAILURE in 6m 07s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/fe2c3913639643d28c8d7b01ac570adc : FAILURE in 8m 38s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/82ca32da15124310b23ad4836c00ea64 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b166546afc97169bab20e907807ae2e4e2d39b45","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-13 12:48:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"15f6eef6179eb47be396a7f941586f65ae71c8a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 13:10:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78e7125d217341398a05f61dd28ee1e3 : SUCCESS in 8m 36s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/9ef7d0f814ff445a9e65e4e932a702ff : SUCCESS in 8m 29s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/f85f960344d44130afd53dde80e1f25d : SUCCESS in 10m 32s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/509c997f2bfc43bab89e1409d84a1599 : SUCCESS in 9m 11s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/c3f92e78e2cf495c9b99d19ce0ff1f77 : SUCCESS in 9m 10s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/5d0f0bda24474e999a351d1f831d28ad : SUCCESS in 5m 37s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/00cb232427ea44fda5b6429b9fb600a8 : FAILURE in 13m 14s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/3e6debf0cdfb4d96b2bb6c0916962a13 : SUCCESS in 14m 30s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/60589eb9b9ab4612a916d5afce42b1cc : SUCCESS in 7m 47s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57c0ca4af1224e63bbbab1b2de40677e : SUCCESS in 3m 27s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/83caf4f0cf5648e0b550cfddf300ff8d : FAILURE in 2m 44s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/1cc8a8e1b29249ba87661f9e3681ae5b : FAILURE in 3m 36s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/217046ce0cbf4cb3b5637c68081fc04e : SUCCESS in 10m 50s","accounts_in_message":[],"_revision_number":2},{"id":"d39ce07874ba1a848b5614357799cb1df058c0cd","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-14 11:59:11.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"84540b002267b82363d13a6dda772e44b4910a77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 12:13:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aedc2a0530c647ceacbc23d11b56fac3 : SUCCESS in 5m 27s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/5e1a86e0723f4961801af8152274eb29 : SUCCESS in 5m 38s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/b619ecc8ba624ec39e1ca628dc03967e : SUCCESS in 4m 59s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/3fc6c0c4079e4db380c4a1d130df9d86 : SUCCESS in 7m 15s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/66fa6cd9e6ad44378e1b5e9f7364bbfe : SUCCESS in 5m 30s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/48d2c0f7cdc6410dbcf6addc1ba27c19 : SUCCESS in 6m 37s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/b6380a6f47e547fdaad4c6c69aef93b7 : FAILURE in 8m 49s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/3b96f549c84f4755a5c73045ebd98e7b : SUCCESS in 10m 05s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/38d3bceaa448420f8f3c9efbff87cbe3 : SUCCESS in 4m 27s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b8843b599944491bae5c90111c27495 : SUCCESS in 3m 52s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/13aedc147e4c41999500535c187bbbaa : SUCCESS in 5m 07s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/f8689dc7ad844a3db1185d2b11557927 : SUCCESS in 5m 42s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/eacb9e4d79f340dbbc7e519d022fa5c8 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":3},{"id":"ce6741170ecfc106ca260db977179adb3f2a3709","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-14 12:31:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6af9b3d257f6f780929b48ea91c5d1c06668140e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 12:50:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8b8ce1883f5424bbe2db5e0ab9a5f4b : SUCCESS in 5m 04s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/604a9d689ad84bad944ccc02bb928d97 : SUCCESS in 5m 31s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fd71dff8ebb46d5b8a038f0e8383e9e : SUCCESS in 5m 02s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/b61092172e1a4aef8a16adc3df037add : SUCCESS in 5m 30s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/8fce7d252d27470e8ceca49dd240a6bc : SUCCESS in 6m 18s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/d500949842234b6aa72208ccb2b7f0a2 : SUCCESS in 6m 08s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/4bad6e0c28344b98b9f588e7afe17fe3 : FAILURE in 7m 57s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/7a851fe4a91b41abaea000116dc748f3 : SUCCESS in 9m 25s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/34b27e6e08e14792960dc39db3e895c7 : SUCCESS in 4m 15s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08b19cbaf89d4d2494b3944a4b2b582e : SUCCESS in 4m 06s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/600d52ab67f7480f9b7d301222597374 : SUCCESS in 5m 58s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/7ce5802bf3934f09ad0f13ee348ef5c6 : SUCCESS in 5m 12s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/26c4b739b93c4d8da73ca9765730fa9e : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":4},{"id":"45d5d4203c417d3c4bb122a3b75ef653bf16b14b","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-14 13:40:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"367a23f88dbf4256e1293eaeaa6919b5455a73f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 13:51:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36f7eb4d46bd429894d9b76b5d044211 : SUCCESS in 5m 14s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/45a1eed162a240ef9a766c8003f844eb : SUCCESS in 5m 14s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2571a5c8be4041248d66aae2bb90c554 : SUCCESS in 5m 09s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/367c4df767f04910947e49d55eef591c : FAILURE in 5m 04s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/eb1590684c23452ca716f89f4dbfe93c : FAILURE in 5m 42s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/6b9b3b87713745fbb2e781fd35ec4a8c : FAILURE in 5m 46s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/917e3108812446449ebfa40b0514433a : FAILURE in 6m 59s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/38e9ad7618cc4fde9c6f1457f6ed605b : SUCCESS in 10m 02s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/39eb2761c91c45eb98f8095ac3796468 : FAILURE in 3m 36s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cec30f26ff6d473086a5c35bc690f3ce : FAILURE in 3m 46s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/5a867c11e2914ecebf7d7ba43602123d : FAILURE in 4m 59s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/8f205329ef4446778cfa1343177f3c78 : FAILURE in 4m 48s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/2b10a2ad523f4387ba9dd9dd41b70753 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":5},{"id":"0eba53f1ea44a4b018761b6fdbf3b15eecfd2432","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-15 16:43:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"db52d46dee2ef427e937f3d0846a42aa28817ddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 16:53:43.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709675765c5c484fb320ce3354cc9708 : SUCCESS in 5m 06s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/5cf5dd6a852a47dea0d13dfff40870f7 : SUCCESS in 5m 36s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd7a84cba1d4e55927f58c68f1d453a : SUCCESS in 4m 26s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/79dd5705ad544036afb7184d131d8267 : FAILURE in 5m 37s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/5ce57dcb23b14e8983aa2e1920bef79f : FAILURE in 5m 04s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/72a651c9b6e94dff9a48eb5e8609b7aa : FAILURE in 5m 20s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/156aba89b8ca4861aed0e0857574b20a : SUCCESS in 7m 47s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/6904f591a3704438bb42d809c47efa00 : SUCCESS in 8m 55s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/1c3b6f1c5c9e4131bf9ca29ec4523c72 : SUCCESS in 5m 12s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b48935ca279f4df6964dcfcb3d96d1b4 : SUCCESS in 3m 39s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/9a9454779b954211b360c677928c6e42 : SUCCESS in 5m 35s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/2845569105914dc286e6af6640f3eeec : SUCCESS in 4m 22s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/91bca10b599643c3a5f80199ba63ee2b : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":6},{"id":"0d76aa85dd1b3d8229a75f96a2697c976a4716c6","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-15 17:03:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1ee8acb32a3b7a6137ba74973409eb10b7cd86e9","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-15 17:06:52.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e5b580a0d102134bf46753a807cafea6bd5d48c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 17:16:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a713440a6e3340169fd53b245c5cfcbe : SUCCESS in 4m 50s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/a8bf6b488610449284e161aba4c59a21 : SUCCESS in 5m 38s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf83ec486e264953a372ac9009aa0435 : SUCCESS in 4m 35s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/8e46ba6fc8424f3abaa2f3904816ad69 : SUCCESS in 4m 27s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/5aadae2b06914c349da44c3997005b9f : SUCCESS in 6m 05s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/46bb61992e034d0fa93c3734b98cf21c : SUCCESS in 5m 33s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/49c3f33e303f4cb683e8ded9e5f4f8b3 : SUCCESS in 7m 39s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/55f94f0545ea4119be58c8cfab536467 : SUCCESS in 8m 38s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b9d9bd79883243bfbf603966004af48d : SUCCESS in 3m 26s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44a2d299fd4245e8bc9ffc6a066c9f0f : SUCCESS in 3m 20s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/ac07f59ed81c47b28b758ba402acecee : SUCCESS in 5m 29s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/75e6b86afb714516805539b8ab95f821 : SUCCESS in 4m 00s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/279fc04d8fae4c1c9a63084c4cb167a3 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":7},{"id":"f636b271d1a984f1973b94baca86186a7ec91168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 17:16:40.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"bf93b629dfa41681084c7960924ceaacd5d37237","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 17:29:16.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c58da8a21d4817951d90df039e5a1f : SUCCESS in 4m 50s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/25729ab459974f12ab823b7d4ff775e8 : SUCCESS in 4m 42s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d0f9eb911034f739c8cf17ae8ad362d : SUCCESS in 4m 27s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/fe8a8b91760b44e287ea956fb5c7a989 : SUCCESS in 4m 42s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/0acf6bf37fa94aca8611ab2803518b1e : SUCCESS in 5m 12s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/1735a3028fdb472697d5b9aa8ca7759f : SUCCESS in 5m 31s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ada891394c6b4e9c92f9b56a3392577a : SUCCESS in 6m 32s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/f4ce593135e04288b3edbbff7b1ff154 : SUCCESS in 8m 29s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/87c2a6bea49a483d99eb457aa01f5f07 : SUCCESS in 3m 24s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b58f036efbc64617b32ff19967e68281 : SUCCESS in 3m 31s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/2e97e791c3ff4ee3be0073da38e4f520 : SUCCESS in 5m 09s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/0ca0d5e02f8c4e04a8d6130e2bc2f318 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":7},{"id":"5e33f4e81a9fe3a317f9cfe727b662fdf43a395a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 17:29:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1f92a18052e9d73957361103b630a65448457ed1","revisions":{"10a1a04e7a420238c9761f0b98be38529238f981":{"kind":"REWORK","_number":1,"created":"2020-02-10 16:28:22.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/96/706896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/96/706896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/96/706896/1"}}},"commit":{"parents":[{"commit":"1e501a6e2175398ffdc506a18e2cb1e1167f8b15","subject":"Stop supporting python \u003c3.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1e501a6e2175398ffdc506a18e2cb1e1167f8b15"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-10 13:56:44.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-10 16:28:11.000000000","tz":120},"subject":"Fix deprecated things","message":"Fix deprecated things\n\n* use inspect.signature instead of inspect.getargspec\n* handle StopIteration properly\n* Make tests.unit.test_hackings runnable at py37\n* Use threading.Thread.is_alive instead of isAlive method\n* Remove usage of six.PY2 and six.PY3\n* Do not use `imp` lib\n\nChange-Id: Ibc901c400e55f201eaaf25eaa1c3eed17b4e0480\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/10a1a04e7a420238c9761f0b98be38529238f981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/10a1a04e7a420238c9761f0b98be38529238f981"}]},"branch":"refs/heads/master"},"b929e0c25467553ba8206fc3c05809e5234d5866":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-13 12:48:07.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/96/706896/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/96/706896/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/96/706896/2"}}},"commit":{"parents":[{"commit":"4f0f394597fd1cb29d93bb1b536a55cbf9b8b8c8","subject":"Rework rally-install-* jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4f0f394597fd1cb29d93bb1b536a55cbf9b8b8c8"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-10 13:56:44.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-13 12:48:07.000000000","tz":0},"subject":"Fix deprecated things","message":"Fix deprecated things\n\n* use inspect.signature instead of inspect.getargspec\n* handle StopIteration properly\n* Make tests.unit.test_hackings runnable at py37\n* Use threading.Thread.is_alive instead of isAlive method\n* Remove usage of six.PY2 and six.PY3\n* Do not use `imp` lib\n\nChange-Id: Ibc901c400e55f201eaaf25eaa1c3eed17b4e0480\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b929e0c25467553ba8206fc3c05809e5234d5866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b929e0c25467553ba8206fc3c05809e5234d5866"}]},"branch":"refs/heads/master"},"9ff4c2fa690a5104fececc3bf503197e771a0811":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-14 11:59:11.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/96/706896/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/96/706896/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/96/706896/3"}}},"commit":{"parents":[{"commit":"000052f67a38579666cc9c446675de148c4f0b4d","subject":"Rework rally-install-* jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/000052f67a38579666cc9c446675de148c4f0b4d"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-10 13:56:44.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-14 11:59:11.000000000","tz":0},"subject":"Fix deprecated things","message":"Fix deprecated things\n\n* use inspect.signature instead of inspect.getargspec\n* handle StopIteration properly\n* Make tests.unit.test_hackings runnable at py37\n* Use threading.Thread.is_alive instead of isAlive method\n* Remove usage of six.PY2 and six.PY3\n* Do not use `imp` lib\n\nChange-Id: Ibc901c400e55f201eaaf25eaa1c3eed17b4e0480\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9ff4c2fa690a5104fececc3bf503197e771a0811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9ff4c2fa690a5104fececc3bf503197e771a0811"}]},"branch":"refs/heads/master"},"ce2388ef9b2b7a3c0b4079d5037d0e3e4f797dd8":{"kind":"REWORK","_number":4,"created":"2020-02-14 12:31:33.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/96/706896/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/96/706896/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/96/706896/4"}}},"commit":{"parents":[{"commit":"000052f67a38579666cc9c446675de148c4f0b4d","subject":"Rework rally-install-* jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/000052f67a38579666cc9c446675de148c4f0b4d"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-10 13:56:44.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-14 12:31:18.000000000","tz":120},"subject":"Fix deprecated things","message":"Fix deprecated things\n\n* use inspect.signature instead of inspect.getargspec\n* handle StopIteration properly\n* Make tests.unit.test_hackings runnable at py37\n* Use threading.Thread.is_alive instead of isAlive method\n* Remove usage of six.PY2 and six.PY3\n* Do not use `imp` lib\n\nChange-Id: Ibc901c400e55f201eaaf25eaa1c3eed17b4e0480\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ce2388ef9b2b7a3c0b4079d5037d0e3e4f797dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ce2388ef9b2b7a3c0b4079d5037d0e3e4f797dd8"}]},"branch":"refs/heads/master"},"864d030b35ca9b607964ff2506031606b4657da9":{"kind":"REWORK","_number":5,"created":"2020-02-14 13:40:09.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/96/706896/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/96/706896/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/96/706896/5"}}},"commit":{"parents":[{"commit":"000052f67a38579666cc9c446675de148c4f0b4d","subject":"Rework rally-install-* jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/000052f67a38579666cc9c446675de148c4f0b4d"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-10 13:56:44.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-14 13:37:36.000000000","tz":120},"subject":"Fix deprecated things","message":"Fix deprecated things\n\n* use inspect.signature instead of inspect.getargspec\n* handle StopIteration properly\n* Make tests.unit.test_hackings runnable at py37\n* Use threading.Thread.is_alive instead of isAlive method\n* Remove usage of six.PY2 and six.PY3\n* Do not use `imp` lib\n\nChange-Id: Ibc901c400e55f201eaaf25eaa1c3eed17b4e0480\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/864d030b35ca9b607964ff2506031606b4657da9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/864d030b35ca9b607964ff2506031606b4657da9"}]},"branch":"refs/heads/master"},"53d8783c34aca7c8893fb16c4b7d77d8b9a565ec":{"kind":"REWORK","_number":6,"created":"2020-02-15 16:43:57.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/96/706896/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/96/706896/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/96/706896/6"}}},"commit":{"parents":[{"commit":"1c45d725e918816df9b5f2348e3a376970668134","subject":"Rework discovery plugins to do not use deprecated imp module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1c45d725e918816df9b5f2348e3a376970668134"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-10 13:56:44.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-15 16:43:48.000000000","tz":120},"subject":"Fix deprecated things","message":"Fix deprecated things\n\n* use inspect.signature instead of inspect.getargspec\n* handle StopIteration properly\n* Make tests.unit.test_hackings runnable at py37\n* Use threading.Thread.is_alive instead of isAlive method\n* Remove usage of six.PY2 and six.PY3\n* Do not use `imp` lib\n\nChange-Id: Ibc901c400e55f201eaaf25eaa1c3eed17b4e0480\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/53d8783c34aca7c8893fb16c4b7d77d8b9a565ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/53d8783c34aca7c8893fb16c4b7d77d8b9a565ec"}]},"branch":"refs/heads/master"},"1f92a18052e9d73957361103b630a65448457ed1":{"kind":"REWORK","_number":7,"created":"2020-02-15 17:03:25.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/96/706896/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/96/706896/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/706896/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/96/706896/7"}}},"commit":{"parents":[{"commit":"1c45d725e918816df9b5f2348e3a376970668134","subject":"Rework discovery plugins to do not use deprecated imp module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1c45d725e918816df9b5f2348e3a376970668134"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-10 13:56:44.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-15 17:02:52.000000000","tz":120},"subject":"Fix deprecated things","message":"Fix deprecated things\n\n* use inspect.signature instead of inspect.getargspec\n* handle StopIteration properly\n* Make tests.unit.test_hackings runnable at py37\n* Use threading.Thread.is_alive instead of isAlive method\n* Remove usage of six.PY2 and six.PY3\n* Do not use `imp` lib\n\nChange-Id: Ibc901c400e55f201eaaf25eaa1c3eed17b4e0480\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1f92a18052e9d73957361103b630a65448457ed1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1f92a18052e9d73957361103b630a65448457ed1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
