)]}'
{"id":"openstack%2Frally~707843","triplet_id":"openstack%2Frally~master~Ibe003c26ef973b18decaa16c03b3589b4002157e","project":"openstack/rally","branch":"master","topic":"py3","hashtags":[],"change_id":"Ibe003c26ef973b18decaa16c03b3589b4002157e","subject":"Fix several issues with plugin discovery error messages","status":"MERGED","created":"2020-02-14 14:10:46.000000000","updated":"2020-02-14 16:16:24.000000000","submitted":"2020-02-14 16:16:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707843-1581696984525-a5ae4dff","meta_rev_id":"0e2c0fe4f1bc5a13e9ce6c202b8649b09fb9f0ff","_number":707843,"virtual_id_number":707843,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-14 16:16:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-02-14 15:54:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-02-14 15:54:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-14 16:16:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84891b491c2c5f65177f772e6f80d3e33b2e51a0","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-14 14:10:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44f5990ce7a10741c88dfad54bc6a5c963ba9ee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 14:35:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c15bc4a0b44ba0b51817e69f409295 : SUCCESS in 6m 18s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/2b897a6a7508402d96966ee998303bbb : SUCCESS in 5m 55s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d7517fe71e8405c978a0445ca848c23 : SUCCESS in 6m 04s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/c3379e35a18f47519342f3b4d6a11c62 : SUCCESS in 6m 51s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/431c11daad7e4df2be4b80f58610596c : SUCCESS in 7m 46s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/898df52dab1a43268d082f41368df902 : SUCCESS in 5m 51s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/dcfae0d0b2ec47de899470fe32e68960 : SUCCESS in 7m 54s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/16d52d0f6a1f42ba8abefa2adfd868db : SUCCESS in 10m 16s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d5c2344f2d964d3d8da98eab05d610ba : SUCCESS in 5m 05s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b28ef11cc064a31a0be55587e5c6abe : SUCCESS in 4m 29s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/b17cfa9afb2b4a5b82127e28afead0cd : SUCCESS in 7m 05s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/01a2d682159940bbab574a865c093e71 : SUCCESS in 5m 38s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/d7cff64fb8e34d83bceee6e5ae7fdf5c : SUCCESS in 9m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2ca1f5490d74e946c9c8ca9ab471f4e03b2bedf7","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-14 15:54:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"99e8d7888d10e84b367de593c0fd2fdfa5791212","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 15:54:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d280297f375ff3f4407b31020ab13eb95ea7be7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 16:16:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"71747b70dc0986ef42771c097d097cd96054d890","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 16:16:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44b601d6fcc40f5a2cdf60d92eb5b99 : SUCCESS in 6m 08s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/bc036e0f54e345a782b55161582bf982 : SUCCESS in 6m 13s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09e0499644a49baa84467a4560d3874 : SUCCESS in 6m 03s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/ac00b6f99ce9463f9ee9ef7048423120 : SUCCESS in 6m 23s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/805b63e27b934ee88b38a7516725e6ea : SUCCESS in 6m 39s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/72d47a9e48f04a1b8a6b64658a6b032e : SUCCESS in 6m 05s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/65ce9b23fa1c4835946c267e63c05fdf : SUCCESS in 9m 13s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/6be53a84970c461b92e44f09353190df : SUCCESS in 10m 38s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/cb1146455d3f47aa82cefb8f3e410e1d : SUCCESS in 4m 18s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4357efb5f1640ae92f8c887da40bd93 : SUCCESS in 5m 14s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/5597959aa03a43f0a6429c6024d70f79 : SUCCESS in 5m 45s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/316d9d215c9d469989f5e14f4e54cdb1 : SUCCESS in 9m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"302216c1a2b3f0fa900eb65343fac95560212682","revisions":{"302216c1a2b3f0fa900eb65343fac95560212682":{"kind":"REWORK","_number":1,"created":"2020-02-14 14:10:46.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/43/707843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/43/707843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/43/707843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/43/707843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/43/707843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/43/707843/1"}}},"commit":{"parents":[{"commit":"000052f67a38579666cc9c446675de148c4f0b4d","subject":"Rework rally-install-* jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/000052f67a38579666cc9c446675de148c4f0b4d"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-14 13:59:29.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-14 14:10:38.000000000","tz":120},"subject":"Fix several issues with plugin discovery error messages","message":"Fix several issues with plugin discovery error messages\n\n* `rally plugin show` should raise errors in case of not found or\n  multiple match\n* remove dumplication of \u0027in\u0027 from PluginNotFound error in case of any\n  platform search\n* handle Scenario plugin not found during task validation\n* doesn\u0027t tag \u0027PluginNotFound\u0027 with \"Unexpected CLI error\" since it is\n  clear error\n\nChange-Id: Ibe003c26ef973b18decaa16c03b3589b4002157e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/302216c1a2b3f0fa900eb65343fac95560212682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/302216c1a2b3f0fa900eb65343fac95560212682"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
