)]}'
{"id":"openstack%2Frally~709704","triplet_id":"openstack%2Frally~master~I68b26ccdaf916b55cb2d27523b145fcadd9a69a8","project":"openstack/rally","branch":"master","topic":"pytest-logs","hashtags":[],"change_id":"I68b26ccdaf916b55cb2d27523b145fcadd9a69a8","subject":"[ci] Register job reports as zuul atrifacts","status":"MERGED","created":"2020-02-25 12:08:26.000000000","updated":"2020-02-25 14:25:00.000000000","submitted":"2020-02-25 14:25:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709704-1582640700236-2a34db55","meta_rev_id":"c1504a0f01f72f0d1dd7f56c083e430db5e676ba","_number":709704,"virtual_id_number":709704,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-25 14:25:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-02-25 14:06:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-02-25 14:06:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 14:25:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"515b7f87bdb10c224fc92a08945d309967425092","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-25 12:08:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebd1b26e6b950193f1d2374c2f3250948755f77d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 12:26:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e12bf6f7afb446e580bf2c70970eb489 : SUCCESS in 5m 33s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/586128628fe64cff96485e529797258e : SUCCESS in 6m 41s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/fda6dd4fbe4b4dd5ad9fcff33422a72c : SUCCESS in 4m 49s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/c061a347d9604cd1b8803ff1ccfec571 : SUCCESS in 5m 30s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/576f7b440fa64252bf11c6109d129f7c : SUCCESS in 6m 51s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/75b7229e53344f3eb0b6e86383822834 : SUCCESS in 6m 11s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/c551147ab77247c58c2cc0f4c28ab081 : SUCCESS in 8m 41s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/50d3ab1fb9ee4390a73042d1f4df67ff : SUCCESS in 6m 14s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/80f402c1ecb84da7a8d07dcb43efdb7a : SUCCESS in 9m 14s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/e9f9abffc3a749f2bf34f9e5c4131e54 : SUCCESS in 5m 28s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1efa391dc50c488281dde27a1143e013 : SUCCESS in 4m 54s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/c258deb7878e4348b715a960e9b442c9 : SUCCESS in 5m 51s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/6e5176210560499aa7d3fb04d186b433 : SUCCESS in 4m 31s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/e652e4e067d44d17bb824f316df5d433 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1b5ccabef61dc1672f145262571a3c5f418f3c9b","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-25 12:47:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"65f6240402f663e5da8856c4afc15025b0e105f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:02:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeb2dd62b5d84ac6a1257b7cabf7c7c9 : SUCCESS in 5m 26s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/16eaf3b0ba4c4ae58d02dcacf7e677dd : POST_FAILURE in 6m 00s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e995ee589eb497a89f5519617e52da6 : POST_FAILURE in 5m 34s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/2ffb0339f39e4b04b3880cd538f57c4f : POST_FAILURE in 5m 41s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/34dfbe60771545998e545e6752721ed5 : POST_FAILURE in 6m 23s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/cd3d4e9a2e5f413b8b726e0b1e8a3278 : SUCCESS in 6m 59s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/6215f257d8ff4ba2a2d84e0b99814408 : POST_FAILURE in 8m 45s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/7448a8c89a8f4d5fa60cab49a5f686f4 : POST_FAILURE in 4m 45s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/6a60d2e6d1d04ecea2f3d3c9d2e3565e : SUCCESS in 9m 01s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/e542fcf8472c482db548dcc883f8dd84 : SUCCESS in 4m 58s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e398743a8b364fad91cbc578e924111a : SUCCESS in 4m 36s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/b6852c5348b24de18a91ff59ae0dd96f : SUCCESS in 5m 28s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/b803cefc1bf943d8a4c6470085b6a0c4 : SUCCESS in 5m 58s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/e93a037971cd4d8e965446b4aeacb0be : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2},{"id":"d28df769a229eba82026d400555a0b88d6abb143","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-25 13:16:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2d9b8c77a5ddc9f7194aec1f073f437aa70caf54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:35:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dcbfbb6edfa453da7b34ba1abc76a33 : SUCCESS in 9m 17s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/3ebc8d548eb741a1af556240ec96fbe9 : SUCCESS in 9m 41s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ee227b6c34c4447b10c4a3b92c7afd8 : SUCCESS in 7m 03s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/9534c4b470f54be78662d3026a5bd724 : SUCCESS in 6m 22s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/c2b3b20c15c04c92bc3fab5fa87e35eb : SUCCESS in 9m 25s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/47fb179fbe144ba799de53f16c88d7a0 : SUCCESS in 7m 08s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/991fadc106e14360ad3558e5b256a74c : SUCCESS in 7m 18s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/32136e5e0ccd415cb1885385c2f03a7f : SUCCESS in 5m 04s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/de64ccc496964755b42a26db22e70c16 : SUCCESS in 12m 15s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f7475a6c4f7b4453a2dfec93eb711650 : SUCCESS in 5m 00s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3cf4a0de39f4995892244ff483a78dc : SUCCESS in 4m 26s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/a6d0a74c391b4320b55aeffef8100573 : SUCCESS in 6m 07s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/0ab17db58fd340319baccd168638f8fc : SUCCESS in 9m 37s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/dc79778683bf40c88f28dc9fef17432c : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":3},{"id":"5d8ee51209bdb73e6e16353465ee09ec449dff84","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-25 14:06:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"11ddbc5ca766898c8fdd92ba36df52859fe6d10d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 14:06:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"06bef595102eaba7ad820a9ee261872970b4bd51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 14:25:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4c113159c3b51bab7b066f2a2270d22bc83c95ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 14:25:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/819ebd0b46984df29e9bc387537afea6 : SUCCESS in 6m 12s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ef25f91952da46ddafa12af68b286e73 : SUCCESS in 5m 06s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6e00e5f6aac4a9ba5fda95695152642 : SUCCESS in 5m 27s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/6ec506e46e0345d18eb03ac12a27271d : SUCCESS in 5m 17s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/55be78fde2c94705b0d899b8c91dac3c : SUCCESS in 5m 36s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/4fe5b43f073947079e9fe9694a181e59 : SUCCESS in 5m 59s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/191b2ad85abe41a784408bd1fa7d79d9 : SUCCESS in 7m 43s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ce94cdc319754f3aa08ddbe778655f38 : SUCCESS in 5m 12s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/b9104fe0309f46668c1d7c7d2ab934ea : SUCCESS in 8m 54s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/285eac34b81e4537beac2479ba5acf48 : SUCCESS in 4m 49s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52daaf42da724b2497c702305b225e4e : SUCCESS in 4m 37s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/6a14fc0cb6f544819a06e34639007f2f : SUCCESS in 5m 48s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/608c04d636644d97a449967b2a4cb964 : SUCCESS in 5m 03s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/399f4951e9ee4596aec7de78b0a3623e : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"89e1cb858800f530b1d51d0d54dc9664d1a1d64b","revisions":{"d0f4afe41a0d9b51a5622e8e64f2413f26c60cec":{"kind":"REWORK","_number":1,"created":"2020-02-25 12:08:26.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/04/709704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/04/709704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/04/709704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/04/709704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/04/709704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/04/709704/1"}}},"commit":{"parents":[{"commit":"1821656002bd938615315e3f61d6cfe0e2acd566","subject":"Remove six usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1821656002bd938615315e3f61d6cfe0e2acd566"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-25 12:07:15.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-25 12:07:15.000000000","tz":120},"subject":"[ci] Register job reports as zuul atrifacts","message":"[ci] Register job reports as zuul atrifacts\n\n* pytest_report.html\n\nChange-Id: I68b26ccdaf916b55cb2d27523b145fcadd9a69a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d0f4afe41a0d9b51a5622e8e64f2413f26c60cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d0f4afe41a0d9b51a5622e8e64f2413f26c60cec"}]},"branch":"refs/heads/master"},"80d46a3e9bce6b513e4d42d6b645a99e0594bea9":{"kind":"REWORK","_number":2,"created":"2020-02-25 12:47:37.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/04/709704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/04/709704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/04/709704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/04/709704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/04/709704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/04/709704/2"}}},"commit":{"parents":[{"commit":"1821656002bd938615315e3f61d6cfe0e2acd566","subject":"Remove six usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1821656002bd938615315e3f61d6cfe0e2acd566"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-25 12:07:15.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-25 12:47:06.000000000","tz":120},"subject":"[ci] Register job reports as zuul atrifacts","message":"[ci] Register job reports as zuul atrifacts\n\nCovers html reports of all rally-tox-* jobs\n\nChange-Id: I68b26ccdaf916b55cb2d27523b145fcadd9a69a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/80d46a3e9bce6b513e4d42d6b645a99e0594bea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/80d46a3e9bce6b513e4d42d6b645a99e0594bea9"}]},"branch":"refs/heads/master"},"89e1cb858800f530b1d51d0d54dc9664d1a1d64b":{"kind":"REWORK","_number":3,"created":"2020-02-25 13:16:26.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/04/709704/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/04/709704/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/04/709704/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/04/709704/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/04/709704/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/04/709704/3"}}},"commit":{"parents":[{"commit":"1821656002bd938615315e3f61d6cfe0e2acd566","subject":"Remove six usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1821656002bd938615315e3f61d6cfe0e2acd566"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-25 12:07:15.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-25 13:16:20.000000000","tz":120},"subject":"[ci] Register job reports as zuul atrifacts","message":"[ci] Register job reports as zuul atrifacts\n\nCovers html reports of all rally-tox-* jobs\n\nChange-Id: I68b26ccdaf916b55cb2d27523b145fcadd9a69a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/89e1cb858800f530b1d51d0d54dc9664d1a1d64b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/89e1cb858800f530b1d51d0d54dc9664d1a1d64b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
