)]}'
{"id":"openstack%2Frally~710432","triplet_id":"openstack%2Frally~master~Ia2cc2dda9090e50fc579dd16e7c7687c951102b6","project":"openstack/rally","branch":"master","topic":"remove_tests","hashtags":[],"change_id":"Ia2cc2dda9090e50fc579dd16e7c7687c951102b6","subject":"Remove unused test helpers","status":"MERGED","created":"2020-02-28 11:00:38.000000000","updated":"2020-02-28 11:32:03.000000000","submitted":"2020-02-28 11:32:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710432-1582889523227-b70934f3","meta_rev_id":"bd9287189935524b7594b90c589d859927d2edab","_number":710432,"virtual_id_number":710432,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 11:32:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-02-28 11:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-02-28 11:10:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-28 11:32:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"833e48c491132ca63bcc58e5a8d9e61767ad8c94","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-28 11:00:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fef11dca021a568dd280dd5cd5ccc83d405836d","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-28 11:10:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3f8ba29c7dd1ad87faf7e3a508176be586048342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 11:16:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca2534d2295f4bc9a7e2fc67ff8b8ceb : SUCCESS in 5m 46s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/e163e05afde44ae589fb5f93d0b5ab25 : SUCCESS in 5m 05s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3125b7ede404633916fb6783a7d7c5f : SUCCESS in 5m 39s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/dfd9b6e461734cc0a27ca1450d5053ce : SUCCESS in 7m 23s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/050bba0bd85145ad96c9e893efd3dae9 : SUCCESS in 7m 04s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/dabb1bb689754b4ca43185f82138dbf0 : SUCCESS in 6m 58s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/c2ae52236f15407cbfa8a4f1b34e1e7b : SUCCESS in 6m 33s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/38512d3e31e34c79af29998cda26d3e9 : SUCCESS in 8m 58s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ae011198368459ea4a4b9b5ccc6075c : SUCCESS in 7m 25s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/5083db0b169f449898ea01304b5c22d7 : SUCCESS in 9m 46s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/c4907c0db8d04ff1b57baaff8336d89c : SUCCESS in 10m 16s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/23d0f38fa22a49c0b6daa1518d78bd1d : SUCCESS in 4m 11s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df6cdaedd06740f188b1f2fcc9300c5c : SUCCESS in 3m 55s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/cf97b9160726496090f7328cd45f2aa0 : SUCCESS in 8m 26s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/671f3d7f237646599cfc55dd30fa33ae : SUCCESS in 5m 23s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/04564740032244cda092e96cd3029992 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"22b283beee4cb15ce5058ca5e3e25c549bbb3b66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 11:16:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69fb345775e1d6565279bfdc88261b0af2349cfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 11:32:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8deac75d9a16a3e41ff91e97f09bcab406a0135b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 11:32:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cf01d0601074d61b68edfef2772d844 : SUCCESS in 6m 19s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/85787a6c5be046f2b681ba2794a1d571 : SUCCESS in 6m 00s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/57fb7b8b5e704690b4c227db0f4f3829 : SUCCESS in 6m 35s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/a77cdb26effa4f5e9ac59f51675ef66c : SUCCESS in 6m 13s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/7ab08f97e8cd4e16b1555ffea538a48e : SUCCESS in 6m 01s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/8cacaa11df5a4da4b4242ac90846d5f3 : SUCCESS in 7m 06s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/65f74bd399fa448ba16ed067ede2b1f0 : SUCCESS in 7m 43s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/f29460cc98094c3cb227d02844802793 : SUCCESS in 10m 11s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/ccf107794c5848778181672598f5a1f1 : SUCCESS in 9m 15s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/bf2f20631a3b4cd89589d6ba3a02a901 : SUCCESS in 5m 35s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da13d3c79fdd4e1c80ae34a7f94ffe07 : SUCCESS in 4m 29s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/a5694de42cc3426983e31e3c2a2fa7d6 : SUCCESS in 5m 36s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/9ff269a8f8bf41369c84a4bddb7b3f26 : SUCCESS in 5m 50s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/afe67d64cdd448f2aa946448c990b083 : SUCCESS in 8m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4687d413bc0b43dda35ff6f003f916bca69956b6","revisions":{"4687d413bc0b43dda35ff6f003f916bca69956b6":{"kind":"REWORK","_number":1,"created":"2020-02-28 11:00:38.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/32/710432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/32/710432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/710432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/710432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/710432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/32/710432/1"}}},"commit":{"parents":[{"commit":"e8cae4b2d04836230f6e76d74cfb128488839635","subject":"Merge \"Ensure python 3.8 is supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e8cae4b2d04836230f6e76d74cfb128488839635"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-28 10:59:49.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-28 10:59:49.000000000","tz":120},"subject":"Remove unused test helpers","message":"Remove unused test helpers\n\nthese helpers left after moving openstack plugins to the separate\nrepository\n\nChange-Id: Ia2cc2dda9090e50fc579dd16e7c7687c951102b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4687d413bc0b43dda35ff6f003f916bca69956b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4687d413bc0b43dda35ff6f003f916bca69956b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
