)]}'
{"id":"openstack%2Frally~713459","triplet_id":"openstack%2Frally~master~Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166","project":"openstack/rally","branch":"master","topic":"docker","hashtags":[],"change_id":"Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166","subject":"Rework docker build","status":"MERGED","created":"2020-03-17 14:38:00.000000000","updated":"2020-03-21 14:03:47.000000000","submitted":"2020-03-21 14:03:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713459-1584799427732-577aa814","meta_rev_id":"16526a283c60bafcbfc0f0a883e6c789a20ed7b9","_number":713459,"virtual_id_number":713459,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-21 14:03:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-03-21 13:48:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-03-21 13:48:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-21 14:03:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"673ac63025592f48c20ce186fed52d28403bfe85","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-17 14:38:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d96ac8193cc77402436417bfa8df8d1bcf89d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 15:03:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/128b26ea2c8a4105aa01838fff9d69c6 : SUCCESS in 6m 00s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/4b78c8c24c504d0e8cbeba6a5809e32c : SUCCESS in 5m 14s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/c80f40f498414be6963dcb781a5fec79 : SUCCESS in 7m 28s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/b1d48807969b4525a4f41f4b6b304813 : FAILURE in 5m 33s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/9e73ecf48b4a4119a268495cafa0f553 : FAILURE in 7m 43s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/c694c4fcdf324636a00bdae175d2a1bd : FAILURE in 7m 42s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/809ca1bbd22e48b69b700368a645f47d : SUCCESS in 8m 03s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/b05d8e53be58417780a7bcca84d84538 : SUCCESS in 9m 10s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/99168cfa568d46838d43b594604eff1b : SUCCESS in 9m 40s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/f1fec28ba12a421aae9b8bb8fce8481e : FAILURE in 7m 18s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/29ef177f2ff149f0bf5bdccc8eebf568 : SUCCESS in 8m 27s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e3b4bac14634f48a775eb236a307d7b : SUCCESS in 4m 40s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/43e9f6cf45a54e498313d151ed45bafd : SUCCESS in 7m 40s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/f3100ea2b746491fa9918466d115add2 : SUCCESS in 5m 01s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/1253948cf2d44830afcea1dca52b6f77 : FAILURE in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5024b5acc138ed2c53d025f8408da3e472c6a087","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-17 16:23:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99c601539665910fdd341fb1d4ef7fe6dc61b18b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 16:37:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa2737187b3c4e129e78c7addceb7478 : SUCCESS in 5m 21s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/896c77620906477d8f12ec25b9fca178 : SUCCESS in 6m 44s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bfff769982e4a5ebd3a299860ae45af : SUCCESS in 6m 01s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/ab145b467ef643109b3d811c92e837b9 : SUCCESS in 7m 55s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/3911610d2eb24b668500bd42c65492a1 : SUCCESS in 7m 29s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/afa43b11ea794f64b8e00d60d3701939 : SUCCESS in 7m 50s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/fc64bcffd39341308d22f9d5e5c7958c : SUCCESS in 7m 22s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/861c09f8c5484b04aca078364760f6c2 : SUCCESS in 9m 13s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bc768c23ce1048daa124fda8770fc242 : SUCCESS in 11m 53s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ddcf98356f6342f794949e32dff51259 : SUCCESS in 11m 25s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/775aea970c4f480cae9127bc94a3dd7b : SUCCESS in 9m 53s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51475dafe2ec476682f1f2915250c4ad : SUCCESS in 4m 48s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/f105260dafa3484ea60e14e7acfcb4a7 : SUCCESS in 6m 59s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/a3da27703c6f49fc89153eff742e111b : SUCCESS in 5m 16s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/96e2bf8b9da54dda8d4d7f1676f6eb2c : FAILURE in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"62945fb9730c674c000c05be3efb27167cd238fa","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-17 16:55:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bb939785066f27eacabcb223ef0d3219580d53ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 17:09:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bdccee0587841ab8ad1e05870069d94 : SUCCESS in 7m 41s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/8db3940fbe5a4f52928a97cf98c66771 : SUCCESS in 4m 33s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c88a48dd14e48a6b7282a8280bead00 : SUCCESS in 5m 15s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/b103c7bd6f784f11a37e0599bc5a2e81 : SUCCESS in 6m 06s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/cadff6d578cc4abd9080a66a77e2e004 : SUCCESS in 5m 15s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/80691725b42945d78180be63123f6a3a : SUCCESS in 4m 04s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/7a6ca2abcf1e445fabd9bf28a0c50a4b : SUCCESS in 5m 15s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/785611b7329a499ca19b8cb86fcf0a6d : SUCCESS in 7m 41s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2d26b725cb2a4186837447a24988dc41 : SUCCESS in 8m 11s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/8f7416edd72d47cba958775aeaacd7d1 : SUCCESS in 8m 05s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/cdd2c2c8b5b34f76bec218ddae6f8898 : SUCCESS in 9m 05s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfe41214a28f458aa361856d35b9fdcb : SUCCESS in 3m 24s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/d6c1e8b0b96741f48b6e62f217a2a438 : SUCCESS in 4m 40s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/028b5524ba454873a777642451d17fdd : SUCCESS in 3m 45s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/30adaf7ec3e242f78716d0fff642a68c : FAILURE in 4m 14s","accounts_in_message":[],"_revision_number":3},{"id":"a533c7e6884922058ff27aed2bf7d9cc2dbc45ab","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 12:13:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b4506adc24e349be2be12526930c7dd14bf688df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:27:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da018c5bb5d4419095602cd9a6f64add : SUCCESS in 4m 38s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/4fdb7c7cec354ed5a5a17215b8bc5f92 : SUCCESS in 4m 15s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/226c46a88bd34eeabcfa8acae339011d : SUCCESS in 4m 14s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/745915d43269498e99dd29679aca72eb : SUCCESS in 5m 30s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/dee2dd0f2e9343fb82594b32eaad324d : SUCCESS in 5m 28s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/aac09bff4584427c9ddc2b0c4cc7dbe5 : SUCCESS in 5m 11s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/814fde54be784e8ba0561e2617177b5e : SUCCESS in 6m 29s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/dbdc8d1e9246413b89ba2d63e342fcc2 : SUCCESS in 6m 35s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9d9d685cf4064871ae70e6daea7e0c93 : SUCCESS in 11m 08s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/390b59d6067a4251a35fd42f6e6007cd : SUCCESS in 8m 37s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/0752ba33808f41afbc907deefc5a796e : SUCCESS in 8m 06s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75dfa3ae386c48b096207079b37a3f16 : FAILURE in 3m 09s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/993a5ddd0a854a1f8de3b5177228eb4b : FAILURE in 4m 51s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/3b4203be97754906988387116be989a1 : FAILURE in 5m 01s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/8668823084894752be3d024a4d1b3669 : FAILURE in 4m 15s","accounts_in_message":[],"_revision_number":4},{"id":"4513138c9a8d171fced0dbb47925666eb739a849","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 12:35:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a9657550c71139f7b782fbb78aa4877af065d4e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:47:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfe0b8be3e10445bbec1428609e4fa15 : SUCCESS in 7m 29s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/cefe096e19dc45289ae33c723a987505 : FAILURE in 3m 48s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/4b36ebc16c784b84b1d00fa01aeb22b7 : FAILURE in 4m 58s","accounts_in_message":[],"_revision_number":5},{"id":"3dce31952d06455014bab9079b8517e634eeb6dd","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 12:49:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0a961947ac677fab1a06593a624f416dccf36b43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:58:09.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71736e177f544b00a423ea081c44bbc2 : SUCCESS in 5m 25s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/dc6d9ac2a90b4fcc835def4c4debedc9 : FAILURE in 3m 59s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/427113cf05784c90bc43dea390cbf76c : FAILURE in 3m 43s","accounts_in_message":[],"_revision_number":6},{"id":"56eb820af22b2f9f851c5d050d0bb7d571b7a3fc","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 13:40:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c388703331af6eb4939e332e11bb79bd297ed009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 13:55:29.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5a0d393abc64efba84806c425c0952a : SUCCESS in 6m 21s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/03b2627fe90e4212ac963acef1c3473d : FAILURE in 9m 16s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/0dc8486179904a7f8d832dc6bfc86b28 : FAILURE in 4m 59s","accounts_in_message":[],"_revision_number":7},{"id":"489133d9aaad4a0c87a03a76e1df105e2763d7b8","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 15:04:23.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b1af904ff71b40ad9c3fdd47ffacb4e1df5f6194","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 15:17:52.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"bde3afd3e62687662836dc5852f9713da30e156c","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 15:29:20.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"129cca02e345c719a19747389f765b8c027141cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 15:39:26.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8713d967b2e47e1b7a9ea3f5e9e485b : SUCCESS in 4m 45s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/9e78f77bbd964168a430ab5815878024 : FAILURE in 7m 41s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/1e0b914e2c39494382e978160106c8e7 : FAILURE in 8m 06s","accounts_in_message":[],"_revision_number":10},{"id":"4a129e382466be9570b82effae68bf22e8773cf2","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 15:46:09.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"b4dd24fe42a968a7ba957375e17f0ea4cfa878d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 16:01:30.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d5cca067fdb427f95013c8feb680b91 : SUCCESS in 4m 32s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/9433a4fa108148b18938586209d724cd : FAILURE in 9m 06s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/79a117dd09f0468da5ce8cf9650ba678 : FAILURE in 3m 58s","accounts_in_message":[],"_revision_number":11},{"id":"8ea4064700a303ff5d722a000c76bf59c6b34968","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 16:07:13.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"8fb78663b74d82bdc08df4dbe6b4884a70fe1375","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 16:08:36.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"1805bb491580dd5913380c8606d6f1c8da9068bd","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 16:26:00.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"cd93f2c508cd52c0ef34f5208857113d91d1b32a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 16:39:15.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cdbefdd2d5740fe81948ba38c0a7f26 : SUCCESS in 4m 59s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/b073fd8d03454dd09907bd4339f5106b : FAILURE in 7m 50s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/fe93044af53a4433b6fc073a918a5295 : FAILURE in 5m 47s","accounts_in_message":[],"_revision_number":14},{"id":"a7d2ed29151f6b6e28f11027f835fb7347beeeb1","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 16:51:56.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"bb39e692797be6ec3a8355801ec74b3fc8b7cbf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 17:04:40.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/058f30465be04692afe9d5efa28f2060 : SUCCESS in 7m 14s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/cdb4753eac6548d081d045a9c282a766 : FAILURE in 7m 56s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/a43406d45d514e8c939f03a7461b2f20 : FAILURE in 6m 47s","accounts_in_message":[],"_revision_number":15},{"id":"dc20e4f07515072c021ccaf8740e9d03e9290a93","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 17:41:01.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"723db15020112564d02a96476a9c51da43173580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 18:01:06.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e22ffc4e27541e7a720a0a54f6e3b3e : SUCCESS in 6m 46s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/27884503bfe344718afe6cfa4aa72af6 : FAILURE in 8m 20s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/74c669fad9d64e4ba231c8f1a2cf5b3d : FAILURE in 9m 05s","accounts_in_message":[],"_revision_number":16},{"id":"b648e5cf2414f6c8ac98d01e5022eab1e9a70618","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 18:43:43.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"90ba6a9360a4523836f78249af3ae0d3c95e1398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 19:13:30.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bd7f081b9ec436985f6aac7e65fb950 : SUCCESS in 6m 18s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/14f779353186440a9cfcb801a5b96ec3 : SUCCESS in 8m 16s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/b3fef3e94ebc4861a139b7afd0c49750 : FAILURE in 8m 22s","accounts_in_message":[],"_revision_number":17},{"id":"82f9d706dbb0329ba9f87a2d3568fa333532de6d","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 19:36:21.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"63b4842c3c586c5482b04c2641c0295764978456","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 19:43:06.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"f68568c72a8b8316a005610fde461e7de6c82ec3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 19:55:23.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f479abc2e37c4da49736ddca41c3a3aa : SUCCESS in 5m 52s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/216bc479764d47c6bc0d3a98174e84ae : FAILURE in 10m 13s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/96ef21f3ad174ac6a0012746e1a87337 : FAILURE in 6m 28s","accounts_in_message":[],"_revision_number":19},{"id":"fa1999f521e3b30651096d38e55674bdd5d60eb2","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 19:55:26.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"4a484ebe787ee337f755ddc2e415ba1f5df4ecca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 20:07:46.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b6c317cf3c4f2cbbcc14e389f0c58e : SUCCESS in 5m 26s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/470b4ae88bbd498faa0d6a887f639a3a : SUCCESS in 10m 18s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/7811bf3ed6c948709d8139b1555c3721 : FAILURE in 8m 42s","accounts_in_message":[],"_revision_number":20},{"id":"a7e943bf500edda6cbb468e68810561aeb31a1c4","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 20:11:29.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"4ca93f46c319165cc585f53c01ccca82edf93b5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 20:25:19.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7e923803f854dfca0bad2b022da81e5 : SUCCESS in 4m 25s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/97fff38db41442768ee50f2af27a7e27 : SUCCESS in 8m 07s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/b1c92f65b77e432094fa512824e00893 : SUCCESS in 7m 45s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/5ab4a47488d54c4d8bafbcf126264f20 : FAILURE in 6m 17s","accounts_in_message":[],"_revision_number":21},{"id":"aae5147169235ba8e4d5d8b75ff199ab94c4be00","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-19 20:25:57.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"21e64b101679900a2c7e933a48f36566c3851026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 20:38:38.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/275f762d889243cd98f96ef4c9fd09d3 : SUCCESS in 4m 26s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/b8b40394ea184bcc83a230a830987443 : SUCCESS in 8m 10s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/1140c6e93937461da0cec5c6c3946d9e : SUCCESS in 7m 52s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/83bd38ed43c2459597ec79a6f2059d3d : SUCCESS in 10m 38s","accounts_in_message":[],"_revision_number":22},{"id":"69920995e657638df899c9a6016ba3ac5bdb5c75","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-20 15:40:09.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"5d77845023bb9f2372721b68466845b50a33d282","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-20 15:40:37.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"093f8ad7ed7d3b65d285ba3ac9f6e78e3e95e174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 15:41:26.000000000","message":"Patch Set 24: Verified-1\n\nUnable to freeze job graph: Pre-review pipeline check does not allow post-review job rally-docker-build","accounts_in_message":[],"_revision_number":24},{"id":"da50f30f976342d5eb94cbe610be12248b4649dc","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-20 15:43:28.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"cc53da34e59db31d22b5b144f8937342bfdd7b9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 15:43:50.000000000","message":"Patch Set 25: Verified-1\n\nUnable to freeze job graph: Pre-review pipeline check does not allow post-review job rally-docker-build-and-push","accounts_in_message":[],"_revision_number":25},{"id":"9541a6f8be39d55175db811134bc652521dc7949","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-20 15:44:52.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"bd9046a135b4c53efa4d336c796f0866072a2d9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:04:56.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/362021201d80428080a1f3815fec87df : SUCCESS in 6m 36s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/9163fed7ed6144b3bcd7f4be81617288 : SUCCESS in 4m 49s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/c15ac3f9ee794c7a9aaa28fd7303ce47 : SUCCESS in 7m 07s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/ddc7a0f8c74a498b8628fb3b709d07f9 : SUCCESS in 8m 32s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/dd40b2764ea247e08d887a17295a22c7 : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":26},{"id":"e084e4d8cb4fa4d1a6b5987e9e79c74cb8efc7a0","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-20 16:26:27.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"2c973ba537129456dc57a7f972d6f6e98f4295ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:26:44.000000000","message":"Patch Set 27: Verified-1\n\nUnable to freeze job graph: Pre-review pipeline check does not allow post-review job rally-docker-build-and-push","accounts_in_message":[],"_revision_number":27},{"id":"d6874e10a7970f33d3e523483b27db32580cde13","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-20 16:32:20.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"ccd4d7fce66ad5f036e3a56f289b16ed5275f09e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:32:37.000000000","message":"Patch Set 28: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/rally on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      post-review: true\n\n  in \"openstack/rally/.zuul.d/zuul.yaml@master\", line 27, column 3","accounts_in_message":[],"_revision_number":28},{"id":"a2445bc960caf9292e9d8390c95f3454e58a2ba4","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-20 16:58:40.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"775553f933d871f31c825dbe35e4c0c78dd8c2bd","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-20 17:04:36.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"e45788bf2589b711131f056fd25ba398a5da450e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 17:22:32.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20848a7a48104ad3b0fc7ed87bde390d : SUCCESS in 6m 12s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/b844d5515071416a959d962550c5cf40 : SUCCESS in 4m 35s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4197a52444048d5b9341eaae7b053c2 : SUCCESS in 5m 22s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/7ed9287864f64367b42adcdfa7efea38 : SUCCESS in 4m 47s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/385539d7dac044cfbfd95cbd868d0beb : SUCCESS in 4m 55s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/ce83721382a744339bac3cafd499c89c : SUCCESS in 5m 30s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/70dfb9d547d94f37892d1d441669802e : SUCCESS in 6m 49s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/b47c1fb70641467fb4d4e8f6f629334e : SUCCESS in 9m 11s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7967c23fc1a245dabadcc3a85a213470 : SUCCESS in 9m 17s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/b478e1448f0c4589bd6f191927f0c061 : SUCCESS in 8m 35s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/f8b9166a41c44365b8fd505de72b2803 : SUCCESS in 9m 58s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b0bccf13c344623b13284c119d1c2ee : SUCCESS in 6m 30s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/bfdeed1fd63a4b5793399aeff3831396 : SUCCESS in 8m 13s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/233e4478d50746cb88e48c22021a9d66 : SUCCESS in 7m 09s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/81c3077aa54b45cf98a13dca6191f4fa : SUCCESS in 10m 39s\n- rally-docker-build-and-push https://zuul.opendev.org/t/openstack/build/90173aacfeea4263b4d82fb1e41e58cc : FAILURE in 7m 06s","accounts_in_message":[],"_revision_number":30},{"id":"b98a35ed7bbeb49f3cfc9bda295de7dc4ec0d39a","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-20 17:32:30.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"17d25aa246ceb505587eeb6d219b311eeb7e3b4e","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-20 17:56:32.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"39d97aacfef9899c98d51efeaac9269f65636dea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 18:21:01.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c48dbd0deb441b8393a62e5712652d : SUCCESS in 4m 36s\n- rally-docker-build-and-push https://zuul.opendev.org/t/openstack/build/66a4ee076d8e4a30bbd3b2aa13e4cfe1 : FAILURE in 12m 02s","accounts_in_message":[],"_revision_number":32},{"id":"d1aff86964f55f82e4cc70d22b51e70b52355d73","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-20 19:06:32.000000000","message":"Patch Set 33: Published edit on patch set 32.","accounts_in_message":[],"_revision_number":33},{"id":"4bbf8ac7a7e131bbb51b1af5ca940fab9150df39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 19:17:33.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b659860775c407eaa173271fca12206 : SUCCESS in 4m 22s\n- rally-docker-build-and-push https://zuul.opendev.org/t/openstack/build/a2c0e014dc414333a103dd6f4105ed20 : FAILURE in 10m 26s","accounts_in_message":[],"_revision_number":33},{"id":"2c526c23ed9b29f5374d7b30e3fcb357c02d0d8e","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-21 10:34:20.000000000","message":"Patch Set 33:\n\nrecheck","accounts_in_message":[],"_revision_number":33},{"id":"b850b30205e2c84f23eb22cbc622c9e1aa3584a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 10:48:47.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a9e77e2d1c458e9fcb5e584cdaca02 : SUCCESS in 4m 17s\n- rally-docker-build-and-push https://zuul.opendev.org/t/openstack/build/8ad5c7754c29421db9f3b6be0cda0dc0 : SUCCESS in 9m 24s","accounts_in_message":[],"_revision_number":33},{"id":"b7c2119bebd985a3f0a03bd1f885ea45d416b82f","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-21 11:00:07.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"8231af2a25e9f19a7e657e18b77ae828b9d18295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 11:28:31.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74da36694cb844129bc9a394e9733d45 : SUCCESS in 6m 43s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/abe0b1a0fea244f9ba2f0408a07cbb0b : SUCCESS in 5m 09s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/573e80c5427a460ebe4d452839d827f0 : SUCCESS in 4m 22s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/3efb4f76a99e4c63878ff89904244171 : SUCCESS in 3m 49s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/3616b2ba824341d687baddf817ee2d50 : SUCCESS in 6m 00s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/953ae724146c4551ac154fecac1441f7 : SUCCESS in 4m 04s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/fc4d7d0aedb34ef9b8d998f188c0c6a6 : SUCCESS in 5m 42s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/659ceb978971474aab1fb2b1d0e51491 : SUCCESS in 8m 01s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3decc44366de45cbb33f96f4ba6b3aaa : FAILURE in 8m 32s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/02061efb6e7b411a8dc71a44595eadfb : SUCCESS in 8m 52s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/6924739ab2c445839840d0f497ac96ef : SUCCESS in 7m 42s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32fccc66c5af4adea86d015a03a944dc : SUCCESS in 6m 23s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/5d210c0cecdc41409b3a82935e7f25c1 : SUCCESS in 7m 27s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/df37afa2386848f29128da8fe27c1388 : SUCCESS in 8m 33s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/268581c7bc3a459db240c4e06871cb32 : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":34},{"id":"9475ecf9f4dcf14733fb372ca600228fba1daa67","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-21 12:03:26.000000000","message":"Patch Set 34: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"facf33727906d2d83687ffd966a626a14bc7c9ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 12:17:02.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9732094aec78476eaf7c4165a93846d7 : SUCCESS in 3m 59s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/9d0ddf1060e6472488ebca4b7f024834 : SUCCESS in 4m 29s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fa470f42bf14409b37b97be91cc94ee : SUCCESS in 3m 05s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/2f6ce258ea434ab98ddd1e12e7bc6d19 : SUCCESS in 5m 02s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/e0a29d2ff73243a3a7c05e8e081b1c12 : SUCCESS in 3m 44s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/efce128c4d044eedb2039e652fa735c1 : SUCCESS in 5m 30s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/67fa2ba5f6fd473589cf94b2fa0f5bb2 : SUCCESS in 5m 35s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/96c84b72c67b466d85d07aef8cdb8f8d : SUCCESS in 6m 50s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/01c4387b8fa34e54bff6c68524b80432 : SUCCESS in 6m 10s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/e3317bc10f504a4293b54d787df24f47 : SUCCESS in 8m 05s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/d2909f7b703a4f6593ec9e1d877a3b6c : SUCCESS in 8m 44s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2617cf9e08b4486aaebee1ec6fb9f4a1 : SUCCESS in 7m 09s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/604475ba146c4df2a8af98a0c97b5641 : SUCCESS in 7m 57s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/704b5ed6b24841efb4c89082680e285e : SUCCESS in 8m 00s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/2cf5c50d38114fc4aff3a5ed462b7daf : SUCCESS in 10m 56s","accounts_in_message":[],"_revision_number":34},{"id":"7efb09ec10309a1250503d389a13651e619bf7c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 12:17:18.000000000","message":"Patch Set 34: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":34},{"id":"fa642766d444be9cc05d3f431346267586edf2c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 12:31:53.000000000","message":"Patch Set 34: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19095a3f78c24219849fa8885ed6d526 : SUCCESS in 3m 43s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/75e2a33ecb8f4c1d9c05d7f2da6092a8 : SUCCESS in 3m 30s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdcda112e5a14b88aa0bbedbfe667d73 : SUCCESS in 3m 14s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/e71b05905fb94dec808a21ae44c2a5cf : SUCCESS in 3m 25s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/7deff3b9c21a4b5eb62f6ec5c9165766 : SUCCESS in 4m 01s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/619f84aafb714164a5a4d06d8ce74be0 : SUCCESS in 3m 58s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/bfd456ba32a947e99066139d6871a276 : SUCCESS in 4m 23s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/1a2a074a9c9c4182aff42235fd82f56b : SUCCESS in 6m 36s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/7b424c2ff6774586ac1315ce47875817 : SUCCESS in 6m 51s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/7d4a48a11e17451bbc51bc37205e5ffb : SUCCESS in 7m 53s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf548c2586df48c9a3bc97abd46c29bc : SUCCESS in 6m 16s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/82145b9c1c1d49e58f944ce8340c752b : SUCCESS in 7m 40s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/1481a987a4fb42f5a2e72fd1fbf28794 : SUCCESS in 6m 55s\n- rally-docker-build-and-push https://zuul.opendev.org/t/openstack/build/cd94a526e8bc4ab2abc8e6c5796ee26f : FAILURE in 8m 24s","accounts_in_message":[],"_revision_number":34},{"id":"5cb30ec99280b06beba97041fcaf2278b3f033a4","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-21 13:37:47.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"ab647df757dd97be27a68fc6b6addbac8f0a6975","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-21 13:48:57.000000000","message":"Patch Set 35: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":35},{"id":"3fe80f3bd794f8e31164ae3949129d795cc4f0f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 13:51:22.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/991ac8a1dd774ff2bb5c23e44e29232d : SUCCESS in 4m 16s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/74e36a8e9ea048d5895a994197d81005 : SUCCESS in 3m 38s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/478bd89931394b0789d7d0ea2edd7f8e : SUCCESS in 3m 37s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/f182b6977bd64d26a641c03c399691fb : SUCCESS in 3m 52s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/53837c29b3c8419ab044d0f9f4e8b6cc : SUCCESS in 4m 02s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/b9edbeb65c684ddda314e8d1c8e73b6a : SUCCESS in 4m 05s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/a06b38b024f7401fb11d2d9b87026bee : SUCCESS in 5m 42s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/b2973207f07a41f694d567164609511e : SUCCESS in 7m 02s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f183504883a04c919aafa48fbd09ab58 : SUCCESS in 6m 41s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/f82b1d4a3ef349ab83caa6ca66c805ee : SUCCESS in 6m 53s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/b45d1423f1d746f890385a5888090b98 : SUCCESS in 7m 54s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddd2c8458b8c455f80fdcef5b3b02c4a : SUCCESS in 6m 04s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/dcf62717b71a4f37ad51b5e11a2ba995 : SUCCESS in 7m 33s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/0d5c3b32c3e446d89010515ca1e9c153 : SUCCESS in 6m 59s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/ba565a567281478b80eb208fed25611e : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":35},{"id":"5dd3f5d39f03e882cd5eaf975922191d44373126","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 13:51:37.000000000","message":"Patch Set 35: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":35},{"id":"d54566784fabf1d48ad98fbd29e41f9263f1723b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 14:03:46.000000000","message":"Patch Set 35: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c98a768aee14a96a9320898652901a2 : SUCCESS in 3m 49s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/dcaf35ffaa02407cbab143befa523d3d : SUCCESS in 3m 50s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/204a4920d4d94e05ac900321dd197001 : SUCCESS in 3m 35s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/901a49b0aa00493aae93a4e6e35edf47 : SUCCESS in 4m 02s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/46c28356f337483e98fc23c603b9081e : SUCCESS in 5m 12s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/c0a4ce71e1764da4a5970f823c0fb036 : SUCCESS in 4m 07s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/72023006c5d54730a1bb5ea6b8a45da8 : SUCCESS in 4m 38s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/af8357c6af3d4a15899c409ec0120191 : SUCCESS in 7m 09s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/e9b71eca1c9a4794a484f463ffd30428 : SUCCESS in 7m 56s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/537edb4d08b14d44a2f919337b912702 : SUCCESS in 8m 42s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91c81316053c4ad79607c4a44412e21c : SUCCESS in 6m 15s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/22f9c075c6ce47939ebcc2d6b8d6e2ba : SUCCESS in 8m 50s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/6ccf739d5cae4f9db2f187713e63e80e : SUCCESS in 7m 47s\n- rally-docker-build-and-push https://zuul.opendev.org/t/openstack/build/c2e3bb5939414677a8ed0833b99c8521 : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":35},{"id":"3581b320c5f90ebbe23478aab3e245f80d02896c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 14:03:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":35}],"current_revision_number":35,"current_revision":"c239b13b27615a115a847a3d4af88a2808e212d9","revisions":{"02424a4dad80a7acba3f48895c96a4889cae8231":{"kind":"REWORK","_number":1,"created":"2020-03-17 14:38:00.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/1"}}},"commit":{"parents":[{"commit":"7d5575d9c27dbfd49eb2c28af2439091712a68f3","subject":"Merge \"Add some tests for verify function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7d5575d9c27dbfd49eb2c28af2439091712a68f3"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-17 14:37:55.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-17 14:37:55.000000000","tz":120},"subject":"WIP: Improve docker","message":"WIP: Improve docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/02424a4dad80a7acba3f48895c96a4889cae8231"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/02424a4dad80a7acba3f48895c96a4889cae8231"}]},"branch":"refs/heads/master"},"d8531371f0a35f014fc29d41db85d85ebf626c72":{"kind":"REWORK","_number":2,"created":"2020-03-17 16:23:28.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/2"}}},"commit":{"parents":[{"commit":"7d5575d9c27dbfd49eb2c28af2439091712a68f3","subject":"Merge \"Add some tests for verify function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7d5575d9c27dbfd49eb2c28af2439091712a68f3"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-17 14:37:55.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-17 16:23:19.000000000","tz":120},"subject":"WIP: Improve docker","message":"WIP: Improve docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d8531371f0a35f014fc29d41db85d85ebf626c72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d8531371f0a35f014fc29d41db85d85ebf626c72"}]},"branch":"refs/heads/master"},"8d058751cba87e080ee9947b5c37cd4b8b28dac5":{"kind":"REWORK","_number":3,"created":"2020-03-17 16:55:28.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/3"}}},"commit":{"parents":[{"commit":"7d5575d9c27dbfd49eb2c28af2439091712a68f3","subject":"Merge \"Add some tests for verify function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7d5575d9c27dbfd49eb2c28af2439091712a68f3"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-17 14:37:55.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-17 16:54:48.000000000","tz":120},"subject":"WIP: Improve docker","message":"WIP: Improve docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8d058751cba87e080ee9947b5c37cd4b8b28dac5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8d058751cba87e080ee9947b5c37cd4b8b28dac5"}]},"branch":"refs/heads/master"},"95a8bfa591dea20ab69e7bc26ba9af8e3464b81f":{"kind":"REWORK","_number":4,"created":"2020-03-19 12:13:15.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/4"}}},"commit":{"parents":[{"commit":"445be80d3f380da14ada42781704ddc894c358eb","subject":"Improme install jobs to run the simlar workload as tox -eself","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/445be80d3f380da14ada42781704ddc894c358eb"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/95a8bfa591dea20ab69e7bc26ba9af8e3464b81f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/95a8bfa591dea20ab69e7bc26ba9af8e3464b81f"}]},"branch":"refs/heads/master"},"a89c7c8155b892aee4427915153f6ee6c369e4ba":{"kind":"REWORK","_number":5,"created":"2020-03-19 12:35:13.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/5"}}},"commit":{"parents":[{"commit":"445be80d3f380da14ada42781704ddc894c358eb","subject":"Improme install jobs to run the simlar workload as tox -eself","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/445be80d3f380da14ada42781704ddc894c358eb"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:35:04.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a89c7c8155b892aee4427915153f6ee6c369e4ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a89c7c8155b892aee4427915153f6ee6c369e4ba"}]},"branch":"refs/heads/master"},"d76ae011bd5aa25e56adbf97411e4f83bf3211cd":{"kind":"REWORK","_number":6,"created":"2020-03-19 12:49:16.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/6"}}},"commit":{"parents":[{"commit":"445be80d3f380da14ada42781704ddc894c358eb","subject":"Improme install jobs to run the simlar workload as tox -eself","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/445be80d3f380da14ada42781704ddc894c358eb"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:49:09.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d76ae011bd5aa25e56adbf97411e4f83bf3211cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d76ae011bd5aa25e56adbf97411e4f83bf3211cd"}]},"branch":"refs/heads/master"},"ae6e11b898157c5d244df2e47aee3a86fc331290":{"kind":"REWORK","_number":7,"created":"2020-03-19 13:40:56.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/7"}}},"commit":{"parents":[{"commit":"445be80d3f380da14ada42781704ddc894c358eb","subject":"Improme install jobs to run the simlar workload as tox -eself","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/445be80d3f380da14ada42781704ddc894c358eb"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 13:40:49.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ae6e11b898157c5d244df2e47aee3a86fc331290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ae6e11b898157c5d244df2e47aee3a86fc331290"}]},"branch":"refs/heads/master"},"e831a0163269e5a78a9423528ad8bf6d03a114df":{"kind":"REWORK","_number":8,"created":"2020-03-19 15:04:23.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/8"}}},"commit":{"parents":[{"commit":"445be80d3f380da14ada42781704ddc894c358eb","subject":"Improme install jobs to run the simlar workload as tox -eself","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/445be80d3f380da14ada42781704ddc894c358eb"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 15:04:15.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e831a0163269e5a78a9423528ad8bf6d03a114df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e831a0163269e5a78a9423528ad8bf6d03a114df"}]},"branch":"refs/heads/master"},"52bb1d1df1c8f0ed3e4e487cefc1acead87892b1":{"kind":"REWORK","_number":9,"created":"2020-03-19 15:17:52.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/9"}}},"commit":{"parents":[{"commit":"445be80d3f380da14ada42781704ddc894c358eb","subject":"Improme install jobs to run the simlar workload as tox -eself","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/445be80d3f380da14ada42781704ddc894c358eb"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 15:17:38.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/52bb1d1df1c8f0ed3e4e487cefc1acead87892b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/52bb1d1df1c8f0ed3e4e487cefc1acead87892b1"}]},"branch":"refs/heads/master"},"d539bcfb2124fbb918d5e97c7d99c4f15ae994e7":{"kind":"REWORK","_number":10,"created":"2020-03-19 15:29:20.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/10"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 15:29:16.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d539bcfb2124fbb918d5e97c7d99c4f15ae994e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d539bcfb2124fbb918d5e97c7d99c4f15ae994e7"}]},"branch":"refs/heads/master"},"309787f4a604c3479f41056c41cb05403a07fbb2":{"kind":"REWORK","_number":11,"created":"2020-03-19 15:46:09.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/11"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 15:46:04.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/309787f4a604c3479f41056c41cb05403a07fbb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/309787f4a604c3479f41056c41cb05403a07fbb2"}]},"branch":"refs/heads/master"},"1ae54b320cc6cedb171214bb9ef73540f159a58d":{"kind":"REWORK","_number":12,"created":"2020-03-19 16:07:13.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/12"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 16:07:08.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1ae54b320cc6cedb171214bb9ef73540f159a58d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1ae54b320cc6cedb171214bb9ef73540f159a58d"}]},"branch":"refs/heads/master"},"fd3c00c5758964b424fca54ab6f9f9c7ea248481":{"kind":"REWORK","_number":13,"created":"2020-03-19 16:08:36.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/13"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 16:08:31.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/fd3c00c5758964b424fca54ab6f9f9c7ea248481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/fd3c00c5758964b424fca54ab6f9f9c7ea248481"}]},"branch":"refs/heads/master"},"1b0b4c7147f9a7d64d84c3878b362af2caa56729":{"kind":"REWORK","_number":14,"created":"2020-03-19 16:26:00.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/14"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 16:20:42.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1b0b4c7147f9a7d64d84c3878b362af2caa56729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1b0b4c7147f9a7d64d84c3878b362af2caa56729"}]},"branch":"refs/heads/master"},"f08346041b218ceb599525dc2a8b6e86dc093780":{"kind":"REWORK","_number":15,"created":"2020-03-19 16:51:56.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/15"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 16:42:21.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f08346041b218ceb599525dc2a8b6e86dc093780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f08346041b218ceb599525dc2a8b6e86dc093780"}]},"branch":"refs/heads/master"},"c4bb536a73e6654ba75de8c6b8b36f8d313da6b1":{"kind":"REWORK","_number":16,"created":"2020-03-19 17:41:01.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/16"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 17:40:55.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c4bb536a73e6654ba75de8c6b8b36f8d313da6b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c4bb536a73e6654ba75de8c6b8b36f8d313da6b1"}]},"branch":"refs/heads/master"},"f9ffbe88fcd480c8f127f8309adf4aba76e774a3":{"kind":"REWORK","_number":17,"created":"2020-03-19 18:43:43.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/17"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 18:43:35.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f9ffbe88fcd480c8f127f8309adf4aba76e774a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f9ffbe88fcd480c8f127f8309adf4aba76e774a3"}]},"branch":"refs/heads/master"},"40f9d09f7293320d6efe7981011b5f1cf1b0365d":{"kind":"REWORK","_number":18,"created":"2020-03-19 19:36:21.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/18"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 19:36:15.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/40f9d09f7293320d6efe7981011b5f1cf1b0365d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/40f9d09f7293320d6efe7981011b5f1cf1b0365d"}]},"branch":"refs/heads/master"},"69ba7cbedf55eee928d269e9724c82f16fe45b06":{"kind":"REWORK","_number":19,"created":"2020-03-19 19:43:06.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/19"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 19:43:01.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/69ba7cbedf55eee928d269e9724c82f16fe45b06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/69ba7cbedf55eee928d269e9724c82f16fe45b06"}]},"branch":"refs/heads/master"},"6417965488e6b7aad6c8574b215233259052d81c":{"kind":"REWORK","_number":20,"created":"2020-03-19 19:55:26.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/20"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 19:55:21.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6417965488e6b7aad6c8574b215233259052d81c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6417965488e6b7aad6c8574b215233259052d81c"}]},"branch":"refs/heads/master"},"95eb7cb1262aa7eb668c885091379ffa5bf57130":{"kind":"REWORK","_number":21,"created":"2020-03-19 20:11:29.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/21"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 20:11:23.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/95eb7cb1262aa7eb668c885091379ffa5bf57130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/95eb7cb1262aa7eb668c885091379ffa5bf57130"}]},"branch":"refs/heads/master"},"0b356222463a2415ba9c97cc85d0dda66331ffef":{"kind":"REWORK","_number":22,"created":"2020-03-19 20:25:57.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/22"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 20:25:50.000000000","tz":120},"subject":"WIP: rework docker","message":"WIP: rework docker\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0b356222463a2415ba9c97cc85d0dda66331ffef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0b356222463a2415ba9c97cc85d0dda66331ffef"}]},"branch":"refs/heads/master"},"abc270c6aba47ab283e9610f79708eeb27c0f5d9":{"kind":"REWORK","_number":23,"created":"2020-03-20 15:40:09.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/23"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-20 15:40:01.000000000","tz":120},"subject":"Rework docker build","message":"Rework docker build\n\n* put Dockerfile at root level and remove all hooks for autobuilds at\n  hub.docker.org\n* run the similar wokrload as `tox -e self` using rally docker image\n* rewrite workload that is used by `tox -e self` at python\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/abc270c6aba47ab283e9610f79708eeb27c0f5d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/abc270c6aba47ab283e9610f79708eeb27c0f5d9"}]},"branch":"refs/heads/master"},"2937cde827cf03b5a7990413945a259db276317e":{"kind":"REWORK","_number":24,"created":"2020-03-20 15:40:37.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/24"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-20 15:40:31.000000000","tz":120},"subject":"Rework docker build","message":"Rework docker build\n\n* put Dockerfile at root level and remove all hooks for autobuilds at\n  hub.docker.org\n* run the similar wokrload as `tox -e self` using rally docker image\n* rewrite workload that is used by `tox -e self` at python\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2937cde827cf03b5a7990413945a259db276317e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2937cde827cf03b5a7990413945a259db276317e"}]},"branch":"refs/heads/master"},"f4bbfd27464fb1fdee015ce902e6b4647e1cb889":{"kind":"REWORK","_number":25,"created":"2020-03-20 15:43:28.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/25"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-20 15:43:16.000000000","tz":120},"subject":"WIP: Rework docker build","message":"WIP: Rework docker build\n\n* put Dockerfile at root level and remove all hooks for autobuilds at\n  hub.docker.org\n* run the similar wokrload as `tox -e self` using rally docker image\n* rewrite workload that is used by `tox -e self` at python\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f4bbfd27464fb1fdee015ce902e6b4647e1cb889"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f4bbfd27464fb1fdee015ce902e6b4647e1cb889"}]},"branch":"refs/heads/master"},"206b8c93714308ca83d936f3e85c95c2574b05d5":{"kind":"REWORK","_number":26,"created":"2020-03-20 15:44:52.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/26"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-20 15:44:46.000000000","tz":120},"subject":"WIP: Rework docker build","message":"WIP: Rework docker build\n\n* put Dockerfile at root level and remove all hooks for autobuilds at\n  hub.docker.org\n* run the similar wokrload as `tox -e self` using rally docker image\n* rewrite workload that is used by `tox -e self` at python\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/206b8c93714308ca83d936f3e85c95c2574b05d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/206b8c93714308ca83d936f3e85c95c2574b05d5"}]},"branch":"refs/heads/master"},"cd325333b1467671be7176de9ab6666c824fa9fe":{"kind":"REWORK","_number":27,"created":"2020-03-20 16:26:27.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/27"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-20 16:26:22.000000000","tz":120},"subject":"WIP: Rework docker build","message":"WIP: Rework docker build\n\n* put Dockerfile at root level and remove all hooks for autobuilds at\n  hub.docker.org\n* run the similar wokrload as `tox -e self` using rally docker image\n* rewrite workload that is used by `tox -e self` at python\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cd325333b1467671be7176de9ab6666c824fa9fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cd325333b1467671be7176de9ab6666c824fa9fe"}]},"branch":"refs/heads/master"},"719772f4f5dbec33ac6a13d73b31f4cfc3f0c549":{"kind":"REWORK","_number":28,"created":"2020-03-20 16:32:20.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/28"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-20 16:32:12.000000000","tz":120},"subject":"WIP: Rework docker build","message":"WIP: Rework docker build\n\n* put Dockerfile at root level and remove all hooks for autobuilds at\n  hub.docker.org\n* run the similar wokrload as `tox -e self` using rally docker image\n* rewrite workload that is used by `tox -e self` at python\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/719772f4f5dbec33ac6a13d73b31f4cfc3f0c549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/719772f4f5dbec33ac6a13d73b31f4cfc3f0c549"}]},"branch":"refs/heads/master"},"8c79a8b6c8bf743afcc9214fe71917f2f491d8ae":{"kind":"REWORK","_number":29,"created":"2020-03-20 16:58:40.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/29"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-20 16:56:42.000000000","tz":120},"subject":"Rework docker build","message":"Rework docker build\n\n* put Dockerfile at root level and remove all hooks for autobuilds at\n  hub.docker.org\n* run the similar wokrload as `tox -e self` using rally docker image\n* rewrite workload that is used by `tox -e self` at python\n* push docker image at gate pipeline\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8c79a8b6c8bf743afcc9214fe71917f2f491d8ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8c79a8b6c8bf743afcc9214fe71917f2f491d8ae"}]},"branch":"refs/heads/master"},"31807af4c6631a50f9078d746ddbdc6cfa4d4d29":{"kind":"REWORK","_number":30,"created":"2020-03-20 17:04:36.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/30"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-20 17:04:28.000000000","tz":120},"subject":"Rework docker build","message":"Rework docker build\n\n* put Dockerfile at root level and remove all hooks for autobuilds at\n  hub.docker.org\n* run the similar wokrload as `tox -e self` using rally docker image\n* rewrite workload that is used by `tox -e self` at python\n* push docker image at gate pipeline\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/31807af4c6631a50f9078d746ddbdc6cfa4d4d29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/31807af4c6631a50f9078d746ddbdc6cfa4d4d29"}]},"branch":"refs/heads/master"},"19d680a0784c4973aca0391c8659d9b20a72adbd":{"kind":"REWORK","_number":31,"created":"2020-03-20 17:32:30.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/31"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-20 17:32:24.000000000","tz":120},"subject":"Rework docker build","message":"Rework docker build\n\n* put Dockerfile at root level and remove all hooks for autobuilds at\n  hub.docker.org\n* run the similar wokrload as `tox -e self` using rally docker image\n* rewrite workload that is used by `tox -e self` at python\n* push docker image at gate pipeline\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/19d680a0784c4973aca0391c8659d9b20a72adbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/19d680a0784c4973aca0391c8659d9b20a72adbd"}]},"branch":"refs/heads/master"},"4165d759ee78b813e3ad9e1cee3ac0679cb5d908":{"kind":"REWORK","_number":32,"created":"2020-03-20 17:56:32.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/32"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-20 17:56:25.000000000","tz":120},"subject":"Rework docker build","message":"Rework docker build\n\n* put Dockerfile at root level and remove all hooks for autobuilds at\n  hub.docker.org\n* run the similar wokrload as `tox -e self` using rally docker image\n* rewrite workload that is used by `tox -e self` at python\n* push docker image at gate pipeline\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4165d759ee78b813e3ad9e1cee3ac0679cb5d908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4165d759ee78b813e3ad9e1cee3ac0679cb5d908"}]},"branch":"refs/heads/master"},"a58419b883cad929b41036383fa0be3df28d4fad":{"kind":"REWORK","_number":33,"created":"2020-03-20 19:06:32.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/33"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-20 19:06:07.000000000","tz":0},"subject":"Rework docker build","message":"Rework docker build\n\n* put Dockerfile at root level and remove all hooks for autobuilds at\n  hub.docker.org\n* run the similar wokrload as `tox -e self` using rally docker image\n* rewrite workload that is used by `tox -e self` at python\n* push docker image at gate pipeline\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a58419b883cad929b41036383fa0be3df28d4fad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a58419b883cad929b41036383fa0be3df28d4fad"}]},"branch":"refs/heads/master"},"9bd6af39fd203de8297f05a28fcde84c75b5f07b":{"kind":"REWORK","_number":34,"created":"2020-03-21 11:00:07.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/34"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-21 10:59:52.000000000","tz":120},"subject":"Rework docker build","message":"Rework docker build\n\n* put Dockerfile at root level and remove all hooks for autobuilds at\n  hub.docker.org\n* run the similar wokrload as `tox -e self` using rally docker image\n* rewrite workload that is used by `tox -e self` at python\n* push docker image at gate pipeline\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9bd6af39fd203de8297f05a28fcde84c75b5f07b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9bd6af39fd203de8297f05a28fcde84c75b5f07b"}]},"branch":"refs/heads/master"},"c239b13b27615a115a847a3d4af88a2808e212d9":{"kind":"REWORK","_number":35,"created":"2020-03-21 13:37:47.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/59/713459/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/59/713459/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/59/713459/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/59/713459/35"}}},"commit":{"parents":[{"commit":"de4fe90b35f4427ece6ea2aabde162376676e792","subject":"Merge \"Improme install jobs to run the simlar workload as tox -eself\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de4fe90b35f4427ece6ea2aabde162376676e792"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-19 12:12:48.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-21 13:37:33.000000000","tz":120},"subject":"Rework docker build","message":"Rework docker build\n\n* put Dockerfile at root level and remove all hooks for autobuilds at\n  hub.docker.org\n* run the similar wokrload as `tox -e self` using rally docker image\n* rewrite workload that is used by `tox -e self` at python\n* push docker image at gate pipeline\n\nChange-Id: Iddd4b3ad518dcf46bf9cb1d67e0b161ad1728166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c239b13b27615a115a847a3d4af88a2808e212d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c239b13b27615a115a847a3d4af88a2808e212d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
