)]}'
{"id":"openstack%2Frally~714278","triplet_id":"openstack%2Frally~master~I9d639d2baa9298a26e7670c5b4b356a8b56583a2","project":"openstack/rally","branch":"master","topic":"deprecate","hashtags":[],"change_id":"I9d639d2baa9298a26e7670c5b4b356a8b56583a2","subject":"Deprecate rally.common.yamlutils and rally.common.fileutils","status":"MERGED","created":"2020-03-21 22:52:39.000000000","updated":"2020-03-22 10:07:56.000000000","submitted":"2020-03-22 10:07:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714278-1584871676966-dfe5b5ea","meta_rev_id":"0e214f9407023b353f7873f352fb848a07dee4ae","_number":714278,"virtual_id_number":714278,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-22 10:07:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-03-22 09:54:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-03-22 09:54:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-22 10:07:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"074438cb0c3206671a95dbdeca3f7645ea466d0d","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-21 22:52:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7edbffaa007182de3f84ebb047618c566661b8a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 23:04:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a74208e9b8104a78920072edb8258496 : SUCCESS in 4m 16s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/0a372b93faa24fe8908366fc45fd279e : SUCCESS in 3m 21s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/b941e14ad2884c0f8f4ea9bb78b27c9f : SUCCESS in 3m 32s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/36c0129dcd4444c1bf429cfaae8b66b0 : SUCCESS in 4m 20s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/0d91a9c819df4b80b5532e4de4bf9e78 : SUCCESS in 4m 01s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/ae928afd7fea498bbf04cfe7de59de0c : SUCCESS in 3m 56s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/dc4ae427a7284a9a94442151eaa2b596 : SUCCESS in 4m 23s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/e2ce166f88ad4111804dcfe567463964 : SUCCESS in 6m 30s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64804e95b24542fa93f781bfdc1d2bde : SUCCESS in 6m 49s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/30b82787900f489ca9af496ceac1a093 : SUCCESS in 6m 45s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/d17655f55c864b17865654578173df17 : SUCCESS in 7m 32s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad7b56fc16b34a188b6a47ab202076e1 : SUCCESS in 5m 33s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/9b8904be06234a7f953ff413bb45b84e : SUCCESS in 7m 16s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/9237ec96c63d484e95bd401a8d446d77 : SUCCESS in 6m 35s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/96f856bc2d2f4747bfab47772c58f768 : SUCCESS in 8m 25s","accounts_in_message":[],"_revision_number":1},{"id":"943f2f3ba135a4e369b7604758c987a92b297817","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-22 09:54:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1806a847361dd9b1ec6c31d07a7e5fc24a9f8e5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 09:54:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37d645d6ad70a405686d583b8e478006156ea141","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 10:07:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5e81d8320c1ff5ccc46a87e8fc443986b0c045b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 10:07:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7831354e54ab41bfa935685c02311b28 : SUCCESS in 4m 19s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/45939a9040504f5dbdb5cd32aeec8a7f : SUCCESS in 4m 35s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/06a1db2fa1e24ecd8f811121f4540f69 : SUCCESS in 3m 40s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/e6bb8ccc04714c129a2d1657089311d5 : SUCCESS in 4m 56s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/682c4d89dab24da1a2c387095b49153b : SUCCESS in 4m 17s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/34b83f3288e1433ca398a49d82fdfdb5 : SUCCESS in 5m 04s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/655611c0fdc740c8a3624607a52909bd : SUCCESS in 5m 20s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/10ad6ac0e3b843e386461acca3ae77fc : SUCCESS in 8m 57s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/79d76350c6e247dd95e8a4308fc20ef6 : SUCCESS in 9m 55s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/8b38979c102e498592f5ced98b34b4e0 : SUCCESS in 7m 46s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3012f0ca03af43cba578572dbd401d4f : SUCCESS in 6m 04s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/4625d4a8bd0d4ccb8add1e1a801f3ae8 : SUCCESS in 9m 31s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/708f92bdc4af4ad1a1472c43f7cca9b0 : SUCCESS in 8m 11s\n- rally-docker-build-and-push https://zuul.opendev.org/t/openstack/build/5e792bf33c974553860ea70c6e451777 : SUCCESS in 12m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30d5a8ed044af7e140d814110d1d834d42ea9e73","revisions":{"30d5a8ed044af7e140d814110d1d834d42ea9e73":{"kind":"REWORK","_number":1,"created":"2020-03-21 22:52:39.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/78/714278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/78/714278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/78/714278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/78/714278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/78/714278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/78/714278/1"}}},"commit":{"parents":[{"commit":"4a466f2bc5d57bdb68ee1055f874ad4cc798b990","subject":"Restruct rally.plugins.common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4a466f2bc5d57bdb68ee1055f874ad4cc798b990"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-21 22:44:08.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-21 22:44:08.000000000","tz":120},"subject":"Deprecate rally.common.yamlutils and rally.common.fileutils","message":"Deprecate rally.common.yamlutils and rally.common.fileutils\n\n* rally.common.fileutils contains stuff that related only to\n  rally.cli.envutils + openstack related helper\n\n* rally.common.yamlutils was designed for CLI purpose and should not be\n  places under rally.common.\n\nChange-Id: I9d639d2baa9298a26e7670c5b4b356a8b56583a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/30d5a8ed044af7e140d814110d1d834d42ea9e73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/30d5a8ed044af7e140d814110d1d834d42ea9e73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
