)]}'
{"id":"openstack%2Frally~714328","triplet_id":"openstack%2Frally~master~I9a4b35643907ce0f101607314aa70222a883368c","project":"openstack/rally","branch":"master","topic":"co_migration","hashtags":[],"change_id":"I9a4b35643907ce0f101607314aa70222a883368c","subject":"[ci] Remove redundant rally task job files","status":"MERGED","created":"2020-03-22 19:17:55.000000000","updated":"2020-03-22 19:56:59.000000000","submitted":"2020-03-22 19:56:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714328-1584907019404-354c625b","meta_rev_id":"f6ee4429db5c66891c87cad3340b43a4e1877719","_number":714328,"virtual_id_number":714328,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-22 19:56:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-03-22 19:26:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-03-22 19:26:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-22 19:56:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5d6e42a5fb9b4b0d53418b31a0e546f4b4f7cb5","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-22 19:17:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e17e0fc5d3ff4402c5cdf8a89725ba384ec115d","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-22 19:24:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98f623d25e69b8bf11d5276a9b5708f9c335108a","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-22 19:26:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"733d2d6e402993531ffd00a62e50bdcaff984784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 19:41:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b7b9578d32a473795a00eacda905b00 : SUCCESS in 4m 14s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/c53442cc9a50485e9792adfeebb8db4d : SUCCESS in 3m 40s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/54f42f2480934c2494e7374113162be2 : SUCCESS in 3m 54s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/f834ab97aa3840838d71eb551517b742 : SUCCESS in 3m 44s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/ce3b3dd0c1224a6e9df29dfa05a903c6 : SUCCESS in 4m 04s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/85c31c4551294f7b8a435e7dcf72d8f6 : SUCCESS in 4m 12s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/38c7ec2a82224d7a9fef80d30d5d73fb : SUCCESS in 5m 12s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/3ea8555a70c149f5b127bdd1aa7f1ff0 : SUCCESS in 6m 31s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0a74b07cc9a44b10b606569efcbc87e7 : SUCCESS in 6m 20s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/4bfeaef4b1a94e0abb4fba6335046bb1 : SUCCESS in 7m 22s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/b52b7644dac34b4d94a9a53d6c8adbf3 : SUCCESS in 7m 36s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3369379427dc481b83a5e765f47772f0 : SUCCESS in 6m 07s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/fad46d262c764d8cb03320ae711d3ab9 : SUCCESS in 7m 45s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/701d9b00494c4ae1b0f5163328ecad44 : SUCCESS in 6m 54s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/c3e85ebdc8784491ab5a54c2a15b8ca2 : SUCCESS in 9m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a86f09dfeb1039bc8e5b7b999084d6548d1db34b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 19:41:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aefec02e8e493c9473be0b342aeb21dc1d96db32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 19:56:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd541c820a94c408117f2dceb1ab238 : SUCCESS in 3m 46s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/467ade60a54c4673aa3d96176d32cfd3 : SUCCESS in 3m 59s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a2312dd7c82468c8fdf775adc0d5b8c : SUCCESS in 3m 42s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/9fae22ed219f42158848d5d13ce9382b : SUCCESS in 4m 53s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/87a8c2744ee041e190e9967978c80d1b : SUCCESS in 5m 32s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/9cf37b9cf8d047efad18e7be58ae1ea6 : SUCCESS in 5m 16s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/bac7c549b27949adbc30e5c94c8a9a26 : SUCCESS in 4m 53s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/98f0105f108a40a7a90b5fae647f2940 : SUCCESS in 6m 48s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/1a2536c1e3e14ccba8e12acdf4d2a279 : SUCCESS in 7m 29s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/66fc3b4843c249ccb7973fe14ebaee2e : SUCCESS in 8m 44s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8bdb3808f7214f079f9382080738b45a : SUCCESS in 6m 35s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/2460dbd473174007ab0a9d39f99a3a30 : SUCCESS in 7m 42s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/e534d8ec0f5f47c38ea7a14d2ef1f86f : SUCCESS in 6m 35s\n- rally-docker-build-and-push https://zuul.opendev.org/t/openstack/build/53323eec575d442e93a38800b6cc6cf4 : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":2},{"id":"56c1c0f0d4a58e62c4a117d8c7c1851c95dff4f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 19:56:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8dbfa86d02cb35e14d597236995926286190fd76","revisions":{"efd2655f60efdddf4b3cdba9175d0e1ca65e61a7":{"kind":"REWORK","_number":1,"created":"2020-03-22 19:17:55.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/28/714328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/28/714328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/714328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/714328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/714328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/28/714328/1"}}},"commit":{"parents":[{"commit":"8dbad440dbd7868361113d11aa09feea8ab205fb","subject":"Deprecate rally task results command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8dbad440dbd7868361113d11aa09feea8ab205fb"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-22 18:49:58.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-22 18:50:00.000000000","tz":120},"subject":"[ci] Remove redundant rally task job files","message":"[ci] Remove redundant rally task job files\n\nThese files left after migration of openstack plugins to the separate\nrepo\n\nChange-Id: I9a4b35643907ce0f101607314aa70222a883368c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/efd2655f60efdddf4b3cdba9175d0e1ca65e61a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/efd2655f60efdddf4b3cdba9175d0e1ca65e61a7"}]},"branch":"refs/heads/master"},"8dbfa86d02cb35e14d597236995926286190fd76":{"kind":"REWORK","_number":2,"created":"2020-03-22 19:24:39.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/28/714328/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/28/714328/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/714328/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/714328/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/714328/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/28/714328/2"}}},"commit":{"parents":[{"commit":"8dbad440dbd7868361113d11aa09feea8ab205fb","subject":"Deprecate rally task results command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8dbad440dbd7868361113d11aa09feea8ab205fb"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-22 18:49:58.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-22 19:23:28.000000000","tz":120},"subject":"[ci] Remove redundant rally task job files","message":"[ci] Remove redundant rally task job files\n\nThese files left after migration of openstack plugins to the separate\nrepo\n\nAlso, this patch fixes unit test for FileExistsValidator validator which\nuses rally-jobs dir for no reasons.\n\nChange-Id: I9a4b35643907ce0f101607314aa70222a883368c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8dbfa86d02cb35e14d597236995926286190fd76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8dbfa86d02cb35e14d597236995926286190fd76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
