)]}'
{"id":"openstack%2Frally~714329","triplet_id":"openstack%2Frally~master~I5ce42221610c8929d8ada5a6f6265997b3be2a99","project":"openstack/rally","branch":"master","topic":"docker_push","hashtags":[],"change_id":"I5ce42221610c8929d8ada5a6f6265997b3be2a99","subject":"Add docker image push job to release pipeline","status":"MERGED","created":"2020-03-22 19:35:35.000000000","updated":"2020-03-22 20:33:09.000000000","submitted":"2020-03-22 20:33:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714329-1584909189358-76a6aaaf","meta_rev_id":"d38b3cf4d7649153692f11c7b09b477185012ba6","_number":714329,"virtual_id_number":714329,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-22 20:33:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-03-22 20:19:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-03-22 20:19:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-22 20:33:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"674bcda09616367054ddd987b652a1e156fe7aa5","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-22 19:35:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb6d97668fbcb1b1908bd2e57c13eecaccdab780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 19:44:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f40c15bd85341a485dc1b9a23825dbe : SUCCESS in 5m 16s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/9e93abfe8f0b4212b6e364c735e54bd5 : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c4bf7411b6da4e02e6d674f5f9a84d5cabd9b444","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-22 19:49:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"de58a81aa08993acadcc135f3ff0f985bb63db1b","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-22 19:57:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6296c7673f0e9539dcd86d36c004cee0e178c396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 20:13:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d7916e406a54e87ae4c750e664958f2 : SUCCESS in 3m 34s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ea6b3694fa8a49e3b8800cc3fe44ea44 : SUCCESS in 3m 24s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7870ccfaa524620914e5dc1ae4b17cf : SUCCESS in 3m 09s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/069527637a9b4f9086d85f1994192c3c : SUCCESS in 3m 31s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/395e7ee0036e4def9c1ee1f3cb7dc7ed : SUCCESS in 4m 02s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/6977da1f168242c1bd640384507cd2f4 : SUCCESS in 3m 49s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/86773d24ece544a3b0361fe336d3cd1a : SUCCESS in 4m 31s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/4b19d299bf2d49fda54a68951206aa52 : SUCCESS in 7m 32s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8d0afc32a69c479584e9cbe5effa6f14 : SUCCESS in 6m 30s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/1afb78b91d4043af82c11c6eb09ab861 : SUCCESS in 6m 44s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/662d309f295546d0b7690e2be4ee871c : SUCCESS in 8m 01s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2850b175fc9141608cf57242866ae695 : SUCCESS in 6m 43s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/bbb0dadb087e458ca7f0bc7b6705c80d : SUCCESS in 7m 12s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/b13d5d510e744336a6816d9cc1a34535 : SUCCESS in 7m 05s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/4b6d0bc8328640a194df0b12c8798568 : SUCCESS in 9m 10s","accounts_in_message":[],"_revision_number":3},{"id":"9bdb71516876840a6ab77955689038acd279f024","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-22 20:19:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"230a090983ddd1178e739cf2cb17b1fc2fbe0ac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 20:19:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"459684919dbd501dae1218ab722fe6d59cf2c9b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 20:33:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4cd72726d445b4f56636f47c73e23664bc59e7c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 20:33:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66c5fdaae9fa4f60865682d9e5f3d3c3 : SUCCESS in 3m 58s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/d4683d14e8a84e6698455e8407d7b04d : SUCCESS in 3m 41s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/62b83a16c70b4a06bf9291357e33c697 : SUCCESS in 3m 43s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/96a2c914d06c412089ae6b4548fd0b1b : SUCCESS in 4m 01s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/5fe17954756a489496ab6caa1a720431 : SUCCESS in 3m 47s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/b23515162ea64e04ac11e87489dfcf0f : SUCCESS in 4m 11s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e87083b907ae467e9e24c252ec56bc46 : SUCCESS in 4m 58s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/6c8834504a01421d887bf521c86b23a1 : SUCCESS in 6m 06s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/bdcd2d27ca0b433c9d0b617d1202b66a : SUCCESS in 6m 43s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/52b654ee51e349a591774d093e5dca0f : SUCCESS in 7m 44s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce34be6ec5974870b37ad3764753b43e : SUCCESS in 5m 50s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/7557c27628574b1ba06b228ec2034b19 : SUCCESS in 7m 56s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/3c16d62fd1f644f0a39512f4488d8023 : SUCCESS in 6m 45s\n- rally-docker-build-and-push https://zuul.opendev.org/t/openstack/build/6ed4b8432eca45288fa3ccc69f2e1d40 : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"57a7b45a06e61e247a0179350f14e2647d0c29e2","revisions":{"a34e2b4320156cafd1ab44fdf73cb7327ce5d17e":{"kind":"REWORK","_number":1,"created":"2020-03-22 19:35:35.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/29/714329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/29/714329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/29/714329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/29/714329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/29/714329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/29/714329/1"}}},"commit":{"parents":[{"commit":"6611305108390f6ca0de2dc634e8b572cb2d0583","subject":"Move task results loader helper to a separate module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6611305108390f6ca0de2dc634e8b572cb2d0583"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-22 19:35:31.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-22 19:35:31.000000000","tz":120},"subject":"WIP: Check docker build job parameters","message":"WIP: Check docker build job parameters\n\nChange-Id: I5ce42221610c8929d8ada5a6f6265997b3be2a99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a34e2b4320156cafd1ab44fdf73cb7327ce5d17e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a34e2b4320156cafd1ab44fdf73cb7327ce5d17e"}]},"branch":"refs/heads/master"},"442eca6a8bdbac56d40b2c2131477b8a9eecc6e9":{"kind":"REWORK","_number":2,"created":"2020-03-22 19:49:02.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/29/714329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/29/714329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/29/714329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/29/714329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/29/714329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/29/714329/2"}}},"commit":{"parents":[{"commit":"6611305108390f6ca0de2dc634e8b572cb2d0583","subject":"Move task results loader helper to a separate module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6611305108390f6ca0de2dc634e8b572cb2d0583"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-22 19:35:31.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-22 19:48:38.000000000","tz":120},"subject":"Add docker image push job to release pipeline","message":"Add docker image push job to release pipeline\n\nChange-Id: I5ce42221610c8929d8ada5a6f6265997b3be2a99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/442eca6a8bdbac56d40b2c2131477b8a9eecc6e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/442eca6a8bdbac56d40b2c2131477b8a9eecc6e9"}]},"branch":"refs/heads/master"},"57a7b45a06e61e247a0179350f14e2647d0c29e2":{"kind":"REWORK","_number":3,"created":"2020-03-22 19:57:20.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/29/714329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/29/714329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/29/714329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/29/714329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/29/714329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/29/714329/3"}}},"commit":{"parents":[{"commit":"6611305108390f6ca0de2dc634e8b572cb2d0583","subject":"Move task results loader helper to a separate module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6611305108390f6ca0de2dc634e8b572cb2d0583"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-22 19:35:31.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-22 19:57:12.000000000","tz":120},"subject":"Add docker image push job to release pipeline","message":"Add docker image push job to release pipeline\n\nChange-Id: I5ce42221610c8929d8ada5a6f6265997b3be2a99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/57a7b45a06e61e247a0179350f14e2647d0c29e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/57a7b45a06e61e247a0179350f14e2647d0c29e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
