)]}'
{"id":"openstack%2Frally~714330","triplet_id":"openstack%2Frally~master~I9ef8d235b8c1e3bd1a5330cb5fe9b99967bbe961","project":"openstack/rally","branch":"master","topic":"docker_push","hashtags":[],"change_id":"I9ef8d235b8c1e3bd1a5330cb5fe9b99967bbe961","subject":"Propose Rally 3.0.0","status":"MERGED","created":"2020-03-22 20:19:27.000000000","updated":"2020-03-23 14:11:13.000000000","submitted":"2020-03-23 14:11:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714330-1584972673141-28344ecc","meta_rev_id":"7ce5f88604ea17fa03521d938a8191a5179651e4","_number":714330,"virtual_id_number":714330,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-23 14:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-03-23 13:55:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-03-23 13:55:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-23 14:11:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d871e9d7799b23ce51f007c3f0196ba8820ad6b3","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-22 20:19:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"767f20b37b059422b9eaa7bbbea0949128aeff3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 20:34:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3938a8ff9c140299646da2146927c16 : SUCCESS in 3m 43s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/d9ef75f92aed45f7babff4658a9e050c : SUCCESS in 3m 54s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/68386f7f4a5f4e73acd948e6b22649c0 : SUCCESS in 3m 30s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/f964b6d254e24038ad9fba154c1e9289 : SUCCESS in 4m 06s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/22ac86d8bf6c49a191fbfbe62b5b1033 : SUCCESS in 4m 29s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/7d54b1a33eb5490aa0622a0521542284 : SUCCESS in 4m 24s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/af3641b4658346f09b90cbadad8b4d1e : SUCCESS in 4m 20s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/3ec8bf6b33ad425abf96ca3ef555c1c3 : SUCCESS in 7m 06s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ce8a4cebec2942e492e3820929208404 : SUCCESS in 7m 12s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/43be5938cb4b43aab9f6bf48042824fe : SUCCESS in 6m 52s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/c7f0d43952784b1f94a50d2673f48334 : SUCCESS in 7m 47s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/090b5b90b09b4c329ef87c579dfc839f : SUCCESS in 5m 56s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/2cd22145af634819ba3bc8800f0385d9 : SUCCESS in 8m 02s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/0f652ebe29fd419b9b7687e1f570ebbc : SUCCESS in 7m 08s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/6840baca53af43cebe8a5fffed355caf : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ffc481d26818406696fb8372d90d261f7c35ec84","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-22 20:42:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"adb31a1473af6698aaaf4ca48e65e975f3857b3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 21:07:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13da2faf4c134c7eab6473f8d2a10d43 : SUCCESS in 7m 28s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/8ffba561abe24cd8800d37f944fb6df3 : SUCCESS in 3m 35s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/452a27dc099d420dbfff6d061c1e8cf0 : SUCCESS in 4m 00s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/bb7ba006f01146c0a76827553e9229c6 : SUCCESS in 7m 57s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/9b300406440b4715bc9cbca2f484993c : SUCCESS in 7m 11s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/32579eddf2bc468ab593c652e9ed871c : SUCCESS in 7m 13s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/057f52eb5d54479ca075f964c679a5fa : SUCCESS in 6m 57s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ecf50420fdd94822990b3614bbeeb046 : SUCCESS in 10m 26s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2c7b6f32ed41470ba9f66561335235f5 : SUCCESS in 11m 58s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/2da64d13dd1a43f3b66a7d5e01ea43c9 : SUCCESS in 6m 40s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/3142ea04e5b843ffbe92730729834055 : SUCCESS in 8m 47s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8eda068f5028405b9948da7091fa72b5 : SUCCESS in 10m 18s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/e866037ce3da4576aa0421f9ccca0752 : SUCCESS in 8m 51s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/52715f222f8e41fd87f7ce09284dd87e : SUCCESS in 9m 42s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/dcb3e92a22874c54ba6f0b0ead45162d : SUCCESS in 15m 10s","accounts_in_message":[],"_revision_number":2},{"id":"3f1b5c7cb0112890fb817c39e881436b08a0822d","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-23 13:55:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"08e6ec17fc356d3c572723e61f770b843abcd777","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 13:57:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6a104f9b04cc07e16ba329e211daebe33e238bb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:11:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb264ff142a4421ea077f170b1b1c124 : SUCCESS in 4m 49s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/978e5a9685ff4e218e1f583f05f84627 : SUCCESS in 3m 51s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec5219406544678a7537641ec76c1b2 : SUCCESS in 3m 37s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/e4ba1a251e51466c890114925fea59a2 : SUCCESS in 3m 57s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/17100e9ef25c4797a28915e7721db084 : SUCCESS in 4m 26s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/29cf79dd77024b86b3e91465f9b2ff9c : SUCCESS in 4m 40s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/7c3a8bf00eec477ea916b0c34ea6af47 : SUCCESS in 6m 03s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/6e48560ca53546819ed7fa6a43a8d0c4 : SUCCESS in 7m 33s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/8c40d1a3fa9b443dae0137bec0830ba5 : SUCCESS in 8m 19s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/19fdce2740ae4555945cd0ae0d7b5673 : SUCCESS in 7m 58s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f06f0b3a915493a88f9cbca65699f56 : SUCCESS in 7m 16s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/3acbfbf40f204a6396f1184eaf7cba78 : SUCCESS in 8m 22s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/f17b61b188944f018a5009d9fea5412c : SUCCESS in 7m 06s\n- rally-docker-build-and-push https://zuul.opendev.org/t/openstack/build/7501f548525b4abc83df67a1d276ac87 : SUCCESS in 9m 40s","accounts_in_message":[],"_revision_number":2},{"id":"bfa5817cabb091e189b3d339ba67967895a311ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:11:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"89662fffe8f77657a0052ae201f659372a8f7c31","revisions":{"15fa18388ff3c6472932a2a6fbcdbeaa0ee7a4a7":{"kind":"REWORK","_number":1,"created":"2020-03-22 20:19:27.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/30/714330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/30/714330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/30/714330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/30/714330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/30/714330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/30/714330/1"}}},"commit":{"parents":[{"commit":"57a7b45a06e61e247a0179350f14e2647d0c29e2","subject":"Add docker image push job to release pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/57a7b45a06e61e247a0179350f14e2647d0c29e2"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-22 20:19:16.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-22 20:19:16.000000000","tz":120},"subject":"Propose Rally 3.0.0","message":"Propose Rally 3.0.0\n\nChange-Id: I9ef8d235b8c1e3bd1a5330cb5fe9b99967bbe961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/15fa18388ff3c6472932a2a6fbcdbeaa0ee7a4a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/15fa18388ff3c6472932a2a6fbcdbeaa0ee7a4a7"}]},"branch":"refs/heads/master"},"89662fffe8f77657a0052ae201f659372a8f7c31":{"kind":"REWORK","_number":2,"created":"2020-03-22 20:42:48.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/30/714330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/30/714330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/30/714330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/30/714330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/30/714330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/30/714330/2"}}},"commit":{"parents":[{"commit":"57a7b45a06e61e247a0179350f14e2647d0c29e2","subject":"Add docker image push job to release pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/57a7b45a06e61e247a0179350f14e2647d0c29e2"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-22 20:19:16.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-22 20:42:39.000000000","tz":120},"subject":"Propose Rally 3.0.0","message":"Propose Rally 3.0.0\n\nChange-Id: I9ef8d235b8c1e3bd1a5330cb5fe9b99967bbe961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/89662fffe8f77657a0052ae201f659372a8f7c31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/89662fffe8f77657a0052ae201f659372a8f7c31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
