)]}'
{"id":"openstack%2Frally~714642","triplet_id":"openstack%2Frally~master~I2d2b50dff1a8739bca8714666acc2efc7369e4d8","project":"openstack/rally","branch":"master","topic":"ci_cover","hashtags":[],"change_id":"I2d2b50dff1a8739bca8714666acc2efc7369e4d8","subject":"Temporary turn off rally-tox-cover job","status":"MERGED","created":"2020-03-24 13:49:32.000000000","updated":"2020-03-24 16:13:21.000000000","submitted":"2020-03-24 16:13:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714642-1585066401894-354c625b","meta_rev_id":"49e4e81caf7bb8973f718fb7564003a90ab4b801","_number":714642,"virtual_id_number":714642,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-24 16:13:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-03-24 13:49:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-03-24 13:49:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-24 16:13:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0d87ccb43dc0506529e53430c48dc7984660ace","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-24 13:49:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43d9e4137b72647bc660caae8898db99be11f192","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-24 13:49:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3fb66d9392cfdfdf442aab7a1f782efc55371120","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 14:14:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca27ee03450d427ba24365b73b96d935 : SUCCESS in 4m 05s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/9d74553530f342838274e9493d8ab3ee : SUCCESS in 4m 38s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b5b168a4ebb452c8fbf6ba8c8ea74c1 : SUCCESS in 3m 55s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/6246c54ea2ec42dfa205323632767555 : SUCCESS in 5m 54s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/a55c16f3143c487d9f4d1a9d333b5958 : SUCCESS in 6m 49s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/0c2722d3d9db4a2faf9f85a08285add3 : SUCCESS in 5m 10s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/79096a386f8f430192d5ce2e50b5da0e : SUCCESS in 6m 44s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/91f9aa249e5849d38792b0a2af0721d0 : SUCCESS in 8m 26s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/8748c1fe3b404325b55754d74f341070 : SUCCESS in 8m 54s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/fb0ad344b68d4ba58b46e863f7d6dc22 : SUCCESS in 9m 21s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/965ad05f7cac4edd8a4daf7392de6170 : SUCCESS in 6m 32s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/36589306c2e4455093622c42cb79b794 : SUCCESS in 9m 21s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/f4e5f4f80e9647f7ad4598f4ef306717 : SUCCESS in 7m 47s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/40d43e1ba8bd4b29a3c4473831af6b36 : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a3caa081124c50390c5c03d4deee52e698d3feee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 14:15:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f5c3b730ccd5862330cfc8d90f1bcdd055cc389d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 14:58:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/422fd40a210c454893e9e2614c9dfd7a : SUCCESS in 7m 46s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/8daa85a029ea4a41beb9e1ad8536668f : SUCCESS in 6m 41s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/d63296a8cb174752b63d3d38cf3f8993 : SUCCESS in 5m 32s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/b318c8089a48407bb9bdb0d85b141535 : SUCCESS in 6m 52s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/bc8b403c76f5415aa9aaf2e08b3b2a2a : SUCCESS in 6m 35s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/c183de085b0440088becfe73feea41dd : SUCCESS in 4m 36s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/d651b63f3d5a4ff9be97bf439bea4d4e : SUCCESS in 10m 06s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/5981957fbbc94e7e82d3c50426e78e96 : SUCCESS in 9m 21s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/4f96bf9dbc0f4fb4acb5316a6c437a93 : FAILURE in 20m 14s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/861076bc4e0443c483e18c55a465c23f : SUCCESS in 6m 36s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/6c6cc51f5c7f4335964bef7e4305f82d : SUCCESS in 8m 45s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/b5362e6f512e42c5ba4a32f721c22408 : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a6d4535788f55e581772407e98e26aa8cdc865f8","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-03-24 15:39:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2eb55736c8f63d439b92b0a0548acfa93b9ba7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 15:57:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b81193ad3964c3d822e22de2ed4ce5f : SUCCESS in 6m 10s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/27a5afbb5e5e4b278947ee8996584c18 : SUCCESS in 4m 48s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6450fc33dfa42118b71741ed0eabafb : SUCCESS in 4m 11s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/89e92c2710a34f21bc0238e0e8d06bd3 : SUCCESS in 4m 08s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/a006406abe154476a412bdc4f935cd8d : SUCCESS in 4m 45s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/8161e6fb866f436ca72c99478885f6ba : SUCCESS in 7m 02s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/71803c06e98e4a86bb37260de4d7f1e8 : SUCCESS in 7m 49s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0d7cdfab14cc46ea897061f1d2dbbfab : SUCCESS in 9m 34s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/78e83d3ec0de4791a67596fd7bf11fee : SUCCESS in 8m 28s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/b3815cac154143a8a00826e7dd6ccd79 : SUCCESS in 10m 18s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4af61cffa340478c9cf78cffc9de350d : SUCCESS in 7m 17s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/04cf77a7b6dd4362a0707ee70447ef45 : SUCCESS in 11m 12s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/45ead3d33c5e4a7ab6d3204319cf1d03 : SUCCESS in 7m 34s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/cd2946529f7e40d39a1c85b55585110b : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c498b6be66e78ff9e43e7278cdfd54feaa7fe002","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 15:57:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"359b9b8c6eed117b8b7d5a676d6120f2170bf69d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 16:13:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a7f279bafff6b63884bdc89e6d0adcb91ade048f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 16:13:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5508f11bc47547a486e925189ba16664 : SUCCESS in 7m 05s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f48be0ca85f74c66a0a63b954218bdac : SUCCESS in 7m 14s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cb95c6f37f74b72ba9b0e0ab3d85395 : SUCCESS in 6m 30s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/76a6f5339c2b4f03a48d8627b67cae20 : SUCCESS in 6m 37s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/1721221ecde14689af23978ef8a77640 : SUCCESS in 9m 07s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/4e7bb16c71fa4d8ebcf943f17f48a938 : SUCCESS in 7m 25s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/4f0cea27873d480d8b29fd7ec47be906 : SUCCESS in 9m 36s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/6700f7f8957b4ddb806433a897163158 : SUCCESS in 11m 02s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/e9cb635c311744cd87b42adadd31e96e : SUCCESS in 9m 28s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c31f6bfdb5da4b8284d5526127215311 : SUCCESS in 8m 17s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/b119a84eeb5d4aa5ac17402cccec68ce : SUCCESS in 10m 10s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/063a0e5189a447b4a326766b87e3b2f9 : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29097cc00f5820e20c53f2cc567fdd5f6e37857f","revisions":{"29097cc00f5820e20c53f2cc567fdd5f6e37857f":{"kind":"REWORK","_number":1,"created":"2020-03-24 13:49:32.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/42/714642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/42/714642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/42/714642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/42/714642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/42/714642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/42/714642/1"}}},"commit":{"parents":[{"commit":"6b86848fc8517ca465cd0caf916168bd312f0a9b","subject":"Merge \"Propose Rally 3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6b86848fc8517ca465cd0caf916168bd312f0a9b"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-24 13:48:15.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-03-24 13:48:15.000000000","tz":120},"subject":"Temporary turn off rally-tox-cover job","message":"Temporary turn off rally-tox-cover job\n\n+ remove unsued jobs\n+ remove mock from test-requirements\n+ move rally-docker-build-and-push from gate queue to post\n\nChange-Id: I2d2b50dff1a8739bca8714666acc2efc7369e4d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/29097cc00f5820e20c53f2cc567fdd5f6e37857f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/29097cc00f5820e20c53f2cc567fdd5f6e37857f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
