)]}'
{"id":"openstack%2Frally~717589","triplet_id":"openstack%2Frally~master~I63b4294cdc87d62c37d5de4c6670dd84bfe2e797","project":"openstack/rally","branch":"master","hashtags":[],"change_id":"I63b4294cdc87d62c37d5de4c6670dd84bfe2e797","subject":"Remove one deprecated warning","status":"ABANDONED","created":"2020-04-05 09:52:10.000000000","updated":"2020-04-27 15:02:37.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f1fbb0425a863e5aa501ed362ed13fc4b73a20ae","_number":717589,"virtual_id_number":717589,"owner":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-05 10:05:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":-1,"date":"2020-04-05 13:54:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2020-04-21 21:05:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-04-05 09:52:54.000000000","_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},{"_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-05 09:52:54.000000000","updated_by":{"_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"},"reviewer":{"_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"},"state":"REVIEWER"},{"updated":"2020-04-05 10:05:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-21 21:05:07.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"19603cfb8ba1db1a7b56549c0c64a3f16c2d8a22","author":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"date":"2020-04-05 09:52:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7f687a75b613d1061f387bf18cf023caf169b93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 10:05:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be33bea492e34789b569137bf2de5145 : SUCCESS in 4m 41s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/ce4149f14c5746f396c09affb1e64d92 : SUCCESS in 6m 23s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/54e471b8dbb640bea6c5a56532751608 : SUCCESS in 5m 37s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9b63411e33a4f9f818ee75a2f82a964 : SUCCESS in 5m 49s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/25eaae970ff1465aa6697240e257f84f : SUCCESS in 6m 01s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/131e4e93610a41558f505dee59a3a707 : SUCCESS in 5m 01s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/d3769b2e7cc34ed7ac4171922fca12dc : SUCCESS in 7m 05s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/6beeb848412441b2ac2e370fe6e8bf1f : SUCCESS in 4m 17s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ea4e0514dcbb4723b20596253d897efa : SUCCESS in 10m 51s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/812fb3a567c44f0180af87c12c09f54b : SUCCESS in 7m 43s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/29675e32da4947b3a173c5bb24375786 : SUCCESS in 9m 31s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/be19dc4fce9649ba8957b98893510f49 : SUCCESS in 7m 33s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05c570674d04423ab7d7b0631d5246f1 : SUCCESS in 8m 05s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/233172c797cc42668207cb9ceb1fc8f7 : SUCCESS in 8m 20s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/cbbf348bfc6a41328682ab4149ebbc61 : SUCCESS in 8m 43s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/c9c01e4af9914ea89659754dd614430b : SUCCESS in 10m 03s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/overview/stories/keystone/authenticate.rst in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/project_info/release_notes/archive/v0.8.0.rst in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/overview/stories/keystone/authenticate.rst in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/feature_request/implemented/LDAP_support.rst in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/project_info/release_notes/archive/v0.11.0.rst in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/overview/stories/keystone/authenticate.rst in the commit\u0027\n  Comments left for invalid file doc/source/quick_start/tutorial/step_5_task_templates.rst\n  Comments left for invalid file doc/source/install_and_upgrade/install.rst\n  Comments left for invalid file doc/source/project_info/release_notes/archive/v0.11.0.rst\n  Comments left for invalid file doc/source/verification/reports.rst\n  Comments left for invalid file doc/source/quick_start/tutorial/step_10_profiling_openstack_internals.rst\n  Comments left for invalid file doc/source/plugins/plugin_reference.rst\n  Comments left for invalid file doc/source/project_info/release_notes/archive/v0.8.0.rst\n  Comments left for invalid file doc/source/quick_start/tutorial/step_2_input_task_format.rst\n  Comments left for invalid file doc/source/quick_start/tutorial/step_4_adding_success_criteria_for_benchmarks.rst\n  Comments left for invalid file doc/source/contribute.rst\n  Comments left for invalid file doc/source/task/index.rst\n  Comments left for invalid file doc/source/feature_request/implemented/LDAP_support.rst\n  Comments left for invalid file doc/source/overview/stories/keystone/authenticate.rst","accounts_in_message":[],"_revision_number":1},{"id":"a5b6912dc1bd1e7ef6ee62737b7aecd20cd5ac66","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-04-05 13:14:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e715461deed0d363038379e47fb6e35d95eaa1b6","author":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"date":"2020-04-05 13:54:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c20bfc8efe400736f8d66210bce6d6009e4e019","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-04-10 13:02:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e045424f0ca121a6665d56133980fbca186fccde","author":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"date":"2020-04-10 14:29:17.000000000","message":"Patch Set 1:\n\nI was joking because you fully wrote the code update in the reviews (thank you for your help!). It would cost you as much as writing the code directly ;)\n\nSorry for the progress here. I have been diving into issues detected by Rally in our gates.\nThe issue when testing OVN seems more one OVN side\nhttps://bugs.launchpad.net/networking-ovn/+bug/1872032\n\nBut I\u0027m not so confident into Rally Swift testing which fails when verifying radosgw (BucketAlreadyExists). I\u0027m considering that concurrency and containers_per_tenant should be equal 1 after checking all params independently. As far as I rember it passed well vs swift reference implementation so I can\u0027t exclude SUT bugs right now.","accounts_in_message":[],"_revision_number":1},{"id":"59573c905d18a04b712b732aec3e6e358086427c","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-04-10 14:43:46.000000000","message":"Patch Set 1:\n\n\u003e I was joking because you fully wrote the code update in the reviews (thank you for your help!). It would cost you as much as writing the code directly ;)\n\nit will cost more time. since it requires to write at least unittests, find all usages and propose deprecation for get_credentials_* methods:) sorry if I took the interesting part if implementation :)\n\n\u003e Sorry for the progress here.\n\nit is ok. I temporary put that warning to ignore list, since functional job fails with it.\n\n\u003e https://bugs.launchpad.net/networking-ovn/+bug/1872032\n\nI guess I saw something like this... hm. Can we try to reproduce this (run the similar scenario) at upstream, so the neutron folks can help you\n\n\u003e But I\u0027m not so confident into Rally Swift testing which fails when verifying radosgw (BucketAlreadyExists). I\u0027m considering that concurrency and containers_per_tenant should be equal 1 after checking all params independently. As far as I rember it passed well vs swift reference implementation so I can\u0027t exclude SUT bugs right now.\n\nhm. we are testing swift scenarios only with containers_per_tenant: 1 ... need to increase this and check","accounts_in_message":[],"_revision_number":1},{"id":"e0200fd315dffbdce5eaa668a67eefc340cd9d30","author":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"date":"2020-04-10 15:11:00.000000000","message":"Patch Set 1:\n\nOur gates differ a little but from the OpenStack ones. We run over and over the same tests vs the same baremetal deployments (we could face with remaining resources). But yes it would make sens to duplicate NeutronNetworks.create_and_delete_subnets with and without routers.\n\nRegarding swift, I find this two possible issues by brute testing. I do verify the internal Rally code to be confident and then exclude rados gw which may be the root cause. But now the new test case runs very well in // (x5).","accounts_in_message":[],"_revision_number":1},{"id":"ccdccbdf1a80828e7876f97ef82ac72bc6483348","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-04-21 21:05:07.000000000","message":"Patch Set 1:\n\nstarted work here - https://review.opendev.org/#/c/721731/","accounts_in_message":[],"_revision_number":1},{"id":"f1fbb0425a863e5aa501ed362ed13fc4b73a20ae","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-04-27 15:02:37.000000000","message":"Abandoned\n\ncode was fixed at rally-openstack side. https://review.opendev.org/#/c/721731/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02baa28e6b23a93dc6e038f94c550a0f9a17cf43","revisions":{"02baa28e6b23a93dc6e038f94c550a0f9a17cf43":{"kind":"REWORK","_number":1,"created":"2020-04-05 09:52:10.000000000","uploader":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"ref":"refs/changes/89/717589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/89/717589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/89/717589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/89/717589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/89/717589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/89/717589/1"}}},"commit":{"parents":[{"commit":"76e89392475e4c69b96149e10eda287a1fea5a90","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/76e89392475e4c69b96149e10eda287a1fea5a90"}]}],"author":{"name":"Cédric Ollivier","email":"ollivier.cedric@gmail.com","date":"2020-04-05 09:51:19.000000000","tz":120},"committer":{"name":"Cédric Ollivier","email":"ollivier.cedric@gmail.com","date":"2020-04-05 09:52:08.000000000","tz":120},"subject":"Remove one deprecated warning","message":"Remove one deprecated warning\n\nhttps://build.opnfv.org/ci/job/functest-opnfv-functest-healthcheck-latest-tempest_smoke-run/789/console\n\nChange-Id: I63b4294cdc87d62c37d5de4c6670dd84bfe2e797\nSigned-off-by: Cédric Ollivier \u003collivier.cedric@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/02baa28e6b23a93dc6e038f94c550a0f9a17cf43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/02baa28e6b23a93dc6e038f94c550a0f9a17cf43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
