)]}'
{"id":"openstack%2Frally~721628","triplet_id":"openstack%2Frally~master~I374c45008dcdfb910d34868df795e6dbf5b962a3","project":"openstack/rally","branch":"master","topic":"bug/1871406","hashtags":[],"change_id":"I374c45008dcdfb910d34868df795e6dbf5b962a3","subject":"Fix parsing of mime-type at subunit-v2-parser","status":"MERGED","created":"2020-04-21 14:24:48.000000000","updated":"2020-04-22 08:03:42.000000000","submitted":"2020-04-22 08:03:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721628-1587542622198-cd7ad812","meta_rev_id":"44b11743100fafd5db9d4ba4e2792976d9489d68","_number":721628,"virtual_id_number":721628,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31877,"name":"Magdalena Orlowska","email":"m.orlowska2@partner.samsung.com"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-22 08:03:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"recommended":{"_account_id":31877,"name":"Magdalena Orlowska","email":"m.orlowska2@partner.samsung.com"},"all":[{"value":1,"date":"2020-04-22 07:29:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31877,"name":"Magdalena Orlowska","email":"m.orlowska2@partner.samsung.com"},{"value":2,"date":"2020-04-22 07:29:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":0,"_account_id":31877,"name":"Magdalena Orlowska","email":"m.orlowska2@partner.samsung.com"},{"value":1,"date":"2020-04-22 07:29:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31877,"name":"Magdalena Orlowska","email":"m.orlowska2@partner.samsung.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 06:14:57.000000000","updated_by":{"_account_id":31877,"name":"Magdalena Orlowska","email":"m.orlowska2@partner.samsung.com"},"reviewer":{"_account_id":31877,"name":"Magdalena Orlowska","email":"m.orlowska2@partner.samsung.com"},"state":"REVIEWER"},{"updated":"2020-04-22 08:03:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86bb4425f75cc0e5de50c11c1ed24fea6f5fab8f","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-04-21 14:24:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"491d97d15eec5cd3b02e1cd1bc6366073e25db0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 14:43:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b3fd605def84154b5072eeb45044b48 : SUCCESS in 6m 55s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/fc95cae4320c4bd393aaaf5ec822347e : SUCCESS in 6m 53s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/c785feba82634eeab9715e7b7ee077fd : SUCCESS in 6m 10s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/605a6329692342bdaf8ceae2f39d211e : SUCCESS in 6m 18s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/a90daeb6d42b4df69b69f421abc0917c : SUCCESS in 7m 02s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/fe586239b4304744a2f3fc7983be887c : SUCCESS in 7m 32s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/53784114ae7e40bbbf9d0dea83a1cd91 : SUCCESS in 6m 42s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/5e61ad06b3ec4431b3a8c71f111c8cd7 : SUCCESS in 6m 30s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/2e0c43970bc64023ac2bf06c2241e248 : SUCCESS in 8m 38s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/be488067b4dd448089422573315acf42 : SUCCESS in 10m 13s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/9bfdec18f2ae405188360e7af5b13a17 : SUCCESS in 11m 00s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/3a6d9f4338f94e3a9e8a882ba10f7f91 : SUCCESS in 9m 51s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/927869307e0245d982f9e7f09576bbe9 : SUCCESS in 9m 19s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/e7a59255002a4c7c8e768d2184b12d34 : SUCCESS in 10m 18s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/02c5215bdce748ec953dd7ba4f825fb2 : SUCCESS in 9m 02s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/f37b17dd5419419cad5306cf639e1ff8 : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"15d27209c8d819a5a0e576ae4bc6ec510407bd29","author":{"_account_id":31877,"name":"Magdalena Orlowska","email":"m.orlowska2@partner.samsung.com"},"date":"2020-04-22 06:14:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ce6ff0f54ed8967e700e97d879f792060908b5f2","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-04-22 07:01:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5928b5d5f73dcdb22f3c4712cfaa1c49385582ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 07:01:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23bd4b9116dd95f37ce5f213a04914f9d51bc575","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 07:24:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/553318e300d54aceba29fe45bf7f440b : SUCCESS in 4m 30s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/57a967580fc448459510196d72204016 : SUCCESS in 6m 28s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/fd2c94acf6ce486aa3bbab5cd1e5339c : SUCCESS in 4m 24s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d343ab175314afe932ca45d5f13253d : SUCCESS in 5m 04s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/603545b56ee9450a9f2a5f4a1428a82b : SUCCESS in 4m 16s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/43927dc4fd7c471c86df3d0298ee83e9 : SUCCESS in 4m 31s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/6eaa982e2c364747a43b0282e23ef38d : SUCCESS in 4m 31s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/e1afc42b1d854d4bbcc401a540bbe5cd : SUCCESS in 6m 33s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/dff91fbefb7d41cc8380829f1897602d : SUCCESS in 7m 45s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/02a1282aab034362a2111f69f3d8ae34 : SUCCESS in 8m 11s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbd51a62a82349bba1f35838560db4d4 : RETRY_LIMIT in 3m 01s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/64b1ee7db48445469d56b9392aaaa2c8 : SUCCESS in 8m 13s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/dce096fe5bc8421799403ba4a7c80c46 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":1},{"id":"11e9ee4e53a63c439ff939f19406ef7279ebd9a9","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-04-22 07:26:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6c45560fb40429cbfac899c938fa834ad846fc05","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-04-22 07:29:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a07f342dc6efe0542596db214f2f02cc4820cc45","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-04-22 07:29:29.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6380d235fb12c3e755e5639cef112c440b285d6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 07:47:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6e571a7ebb4f099453afdaa04b0a83 : SUCCESS in 5m 10s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b4baeb56564b4834b6a18fe85beb4472 : SUCCESS in 6m 21s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/a6ea10ff1f3240a9b2b1b44b499f4b47 : SUCCESS in 5m 01s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc63771d9594806955dd94b78803a4b : SUCCESS in 5m 03s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/ebd2aad0a2584fbba8cbef6e91af4cad : SUCCESS in 5m 04s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/007fe184a5ef4e519d1ac8af519b1ab8 : SUCCESS in 5m 16s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/01959665ef9f436d9826d53ab1c8cadd : SUCCESS in 5m 37s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/353fea7a7b6a42dea80ead8fa27aafc7 : SUCCESS in 7m 16s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/92e6db85c7fb4c7fb10712e917959ece : SUCCESS in 12m 42s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7f1a01b51d794707aecb29f9cb64c848 : SUCCESS in 10m 48s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/45eea4c5ec4c4af8a4bc487c1b068d11 : SUCCESS in 8m 52s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/112a06df9c7d44bbbb25c2f96b064c76 : SUCCESS in 9m 38s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11e36a4a32ed41bb8f0cf191bc08b78f : SUCCESS in 9m 11s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/ffaf87b27e134685ae431daa2c8b0e8f : SUCCESS in 8m 52s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/e2e674b7a18448eb9232ccf569b36951 : SUCCESS in 8m 37s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/5ea7a9f2f0324dabb9bef30c3ea1c38f : SUCCESS in 12m 02s","accounts_in_message":[],"_revision_number":2},{"id":"b33fef24a67f0b3b4a31b6ff5cb2dbf4ebb5783f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 07:47:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d2783446d24d7ef5a1706e67e7e1f3614c7bd98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 08:03:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c141dbbb9540719b30a932b821d8f4 : SUCCESS in 5m 28s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/20386238e22c4241b5418824c61347b5 : SUCCESS in 6m 01s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/cc101d7265f14afd8195c352d84e844f : SUCCESS in 4m 39s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3215604bec24f4c9d516b47e7ca5e60 : SUCCESS in 5m 04s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/be8a2b961b8540b589ae0baf6e37c655 : SUCCESS in 5m 46s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/e4813a1f7ceb4dc29db44fea9c1d68dc : SUCCESS in 6m 13s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/6d39203a19ee4bb18bd075f884b1d3b7 : SUCCESS in 4m 51s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/48de4fb398e64b019191bfc778df30ec : SUCCESS in 6m 58s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/4d89f57cc5f44877b2178949875b5719 : SUCCESS in 8m 48s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/9df7b73acdc94bbe9f154799d830a0ec : SUCCESS in 7m 50s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6543f5b66314657a5ca86c3062a3c6c : SUCCESS in 8m 43s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/1d4ed8e9e6b54ddfa1c2a732cb58bbb3 : SUCCESS in 8m 57s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/669864c9ca734de89a431310f6b7d42d : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":2},{"id":"1a88c53f89f8918668b24c1b5b9c33364551fe50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 08:03:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b9c303c2f47b4cf38393372bb0310fbf87147b50","revisions":{"05f2ee390c9ec2ed9560e147b28d4a35b8612101":{"kind":"REWORK","_number":1,"created":"2020-04-21 14:24:48.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/28/721628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/28/721628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/721628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/721628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/721628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/28/721628/1"}}},"commit":{"parents":[{"commit":"f1552e9f0a2a2e8926cde20e700f1ee89091c8fe","subject":"Improve rally-docker jobs to be usefull for rally-openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f1552e9f0a2a2e8926cde20e700f1ee89091c8fe"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-04-21 14:23:31.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-04-21 14:23:31.000000000","tz":180},"subject":"Fix parsing of mime-type at subunit-v2-parser","message":"Fix parsing of mime-type at subunit-v2-parser\n\n+ ignore unwanted test id with strange data\n+ do not try to split charset from mime-type. we do not use it.\n\nChange-Id: I374c45008dcdfb910d34868df795e6dbf5b962a3\nCloses-Bug: #1871406\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/05f2ee390c9ec2ed9560e147b28d4a35b8612101"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/05f2ee390c9ec2ed9560e147b28d4a35b8612101"}]},"branch":"refs/heads/master"},"b9c303c2f47b4cf38393372bb0310fbf87147b50":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-22 07:29:24.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/28/721628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/28/721628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/721628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/721628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/721628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/28/721628/2"}}},"commit":{"parents":[{"commit":"d6cb04faef10ec1c398ed5eeeb4359664db07ebd","subject":"[ci] Fix install script for ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d6cb04faef10ec1c398ed5eeeb4359664db07ebd"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-04-21 14:23:31.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-04-22 07:29:24.000000000","tz":0},"subject":"Fix parsing of mime-type at subunit-v2-parser","message":"Fix parsing of mime-type at subunit-v2-parser\n\n+ ignore unwanted test id with strange data\n+ do not try to split charset from mime-type. we do not use it.\n\nChange-Id: I374c45008dcdfb910d34868df795e6dbf5b962a3\nCloses-Bug: #1871406\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b9c303c2f47b4cf38393372bb0310fbf87147b50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b9c303c2f47b4cf38393372bb0310fbf87147b50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
