)]}'
{"id":"openstack%2Frally~721783","triplet_id":"openstack%2Frally~master~I57eeeffc22c7da4c735467dc5b5af745dc3fbd55","project":"openstack/rally","branch":"master","topic":"install_script","hashtags":[],"change_id":"I57eeeffc22c7da4c735467dc5b5af745dc3fbd55","subject":"[ci] Fix install script for ubuntu","status":"MERGED","created":"2020-04-22 07:28:56.000000000","updated":"2020-04-22 08:02:34.000000000","submitted":"2020-04-22 08:02:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721783-1587542554191-6b5ca73b","meta_rev_id":"2b6dfac5a4ac36faf660d24ef6d5248849fde012","_number":721783,"virtual_id_number":721783,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-22 08:02:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-04-22 07:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-04-22 07:29:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 08:02:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"410de957338cf8f433a7769de5d4ef2bcf3f7408","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-04-22 07:28:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"777af8b4b25b8a77e38ca19a4c4fbe6b325405ea","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-04-22 07:29:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b88f3b9ce4461c1ce9223bfd4d8bbad01e5027c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 07:47:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b27d3f1352764d59a3f8fe52e8767e01 : SUCCESS in 7m 40s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/609edf7ed7c545138960ebf36e019cfe : SUCCESS in 7m 36s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/91a8f5acbe99472da355db58db079674 : SUCCESS in 6m 03s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6df4092b8794e3d89941ad1d497eb86 : SUCCESS in 5m 14s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/8435b7aff048467d9c384948c766da89 : SUCCESS in 6m 01s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/0c09afef335d440ca3352baf67096d9a : SUCCESS in 4m 30s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/0805dfcc76754b63a2767ad3476e4e88 : SUCCESS in 5m 31s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/690826517be94d3e8257d95b8edc1303 : SUCCESS in 4m 31s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ab51ec69d19742c3a099b5364d68c1fc : SUCCESS in 8m 38s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5b5281164ee0490a8e695d4cb9eacd2e : SUCCESS in 7m 20s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/dd8751950a124fc09b4277b9219c1cb7 : SUCCESS in 8m 39s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/61445a0e45094763a38acba3e33a4653 : SUCCESS in 10m 28s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b17c7b64a1c44b829f19cf9e2ca1c18a : SUCCESS in 7m 40s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/d710957a0eac4a7897d8478d038f6c01 : SUCCESS in 9m 21s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/d207360859874a17878c2dfffd49feac : SUCCESS in 8m 12s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/c5b352bd62044fb2a17e7eeb247f0ea0 : SUCCESS in 9m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8c1df8cedc4630472e3b9e5d1e4fbafb22486046","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 07:47:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03d0b6e12e4d427532e3bbed12dd1bc1d984d955","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 08:02:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4de7a04946a542f0b2770a654a6eb0db : SUCCESS in 5m 14s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/0224c8d1df274ab4a2dfd210bf9f8aab : SUCCESS in 5m 22s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/dddd1689487d4023b491464bab600625 : SUCCESS in 5m 11s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba0d29bc186b4a96a710fec29a97ce47 : SUCCESS in 4m 04s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/9511efacee9d414b8ccee4fe8b9c35d9 : SUCCESS in 4m 23s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/3dd3cecc8a3d4d22ba4c182cad2d95f9 : SUCCESS in 5m 33s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/ff8ee06df00f4dcf8937fc9eb2cc86dd : SUCCESS in 5m 32s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/89bfeaebb8be4560ab893cb420b48151 : SUCCESS in 8m 50s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/4a61cf7d789746acae238786b94212a4 : SUCCESS in 8m 36s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/ee834da46be745699754c11b6d7593d9 : SUCCESS in 8m 22s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2c5168ef9a1439cb6e0bb9a57e7ca1a : SUCCESS in 7m 40s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/fa7d0d66e3ac4d479eb82683d531fb93 : SUCCESS in 7m 50s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/af9e6048aa0b4419848c0883b8943d8c : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":1},{"id":"53d99bed791ddd584c4277f206afdbea992e066a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 08:02:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6cb04faef10ec1c398ed5eeeb4359664db07ebd","revisions":{"d6cb04faef10ec1c398ed5eeeb4359664db07ebd":{"kind":"REWORK","_number":1,"created":"2020-04-22 07:28:56.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/83/721783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/83/721783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/83/721783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/83/721783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/83/721783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/83/721783/1"}}},"commit":{"parents":[{"commit":"f1552e9f0a2a2e8926cde20e700f1ee89091c8fe","subject":"Improve rally-docker jobs to be usefull for rally-openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f1552e9f0a2a2e8926cde20e700f1ee89091c8fe"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-04-22 07:28:51.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-04-22 07:28:51.000000000","tz":180},"subject":"[ci] Fix install script for ubuntu","message":"[ci] Fix install script for ubuntu\n\nChange-Id: I57eeeffc22c7da4c735467dc5b5af745dc3fbd55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d6cb04faef10ec1c398ed5eeeb4359664db07ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d6cb04faef10ec1c398ed5eeeb4359664db07ebd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
