)]}'
{"id":"openstack%2Frally~726117","triplet_id":"openstack%2Frally~master~I516d2bb46091a607122036d46ea03e7ae3ba0a67","project":"openstack/rally","branch":"master","topic":"validation","hashtags":[],"change_id":"I516d2bb46091a607122036d46ea03e7ae3ba0a67","subject":"Do not inialize scenario cls at scenario argument validator","status":"MERGED","created":"2020-05-07 13:29:17.000000000","updated":"2020-05-07 14:06:19.000000000","submitted":"2020-05-07 14:06:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726117-1588860379092-92884968","meta_rev_id":"36f281a491685f70cf3dec1b4e1391703c73c045","_number":726117,"virtual_id_number":726117,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 14:06:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-05-07 13:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-05-07 13:30:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 14:06:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6248b4a53f3acfe19f4d9af7ed0ae8e86d0d9f28","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-05-07 13:29:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"104ba273754306a2505dc4207619289063205b13","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-05-07 13:30:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"019501c50b4b1061147f37c88e8ba2a92eda2465","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:51:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c683aa2f008b4572812be79421585e96 : SUCCESS in 6m 44s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/c4ee8d98d2904d8eacaa67a10a4144db : SUCCESS in 6m 57s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/29083804a4764faebaa3ddfc8bac15b6 : SUCCESS in 6m 00s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/6655dcaee0204d5eaf5c9faf05cf3e31 : SUCCESS in 4m 02s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/f14656642a434eec85c58bec9ef2ee82 : SUCCESS in 6m 44s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/103bba702f5349fa8c0c9a46f3558dbb : SUCCESS in 5m 20s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/8ef57891eb614212ad576a849a526163 : SUCCESS in 5m 01s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/9a45afbe0e68464da3a72d7a86e56302 : SUCCESS in 6m 55s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/68437855e0d4424787fb9ee30ac0bdaf : SUCCESS in 7m 48s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9e31b323b6f84389987f0624be7343eb : SUCCESS in 8m 21s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/9f225239902e4a3cb2ce4a1f7bd540de : SUCCESS in 7m 32s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/170e3dc402eb4c7eb1baf02909cc49b7 : SUCCESS in 6m 15s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/dd7c7992a7c345419cb27a9c8d2bc6b3 : SUCCESS in 10m 04s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/2482ec388e08404da75fb8a2089bb902 : SUCCESS in 7m 23s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/7beecdb32c564e98ac8bf0254973cd77 : SUCCESS in 12m 21s","accounts_in_message":[],"_revision_number":1},{"id":"afad47e4eea84e556f4238d18df0a4021be3ef50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:51:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35e93df7bcd880ec13f526020ae101c999bb1799","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:06:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a5874c7db54535bf4858143eced9fc : SUCCESS in 12m 36s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/77eaba537daa4e58a68bcad9cfbbe9ce : SUCCESS in 12m 30s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/d6e10cac95a0475ab196f023c6dc3142 : SUCCESS in 6m 43s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ad46f8f289d4616b8ec79c3e81da6d8 : SUCCESS in 5m 18s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/28b94da1aced40b4a6585f88c8d87260 : SUCCESS in 4m 32s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/5428dd80e41949bfa158dc157454520f : SUCCESS in 4m 47s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/c5a4934f466b4cba9ca3300dc35f2b35 : SUCCESS in 9m 24s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/d721b21f4aea4e7eba2ebec0035e3629 : SUCCESS in 9m 32s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/a781840c713640069fb4c663dc8712c6 : SUCCESS in 10m 49s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e0a9039d7a14a8785770597fe3f8118 : SUCCESS in 9m 01s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/3109c08c6113454e95e1209b4b8be640 : SUCCESS in 9m 33s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/0a69a12facfb463abaede9928c4b719c : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8446e9bebd2ba05d7d1f6c0b25b10ca953474973","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:06:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"166435fb4feea5f4d9bfac59e79d0c8ffeec6291","revisions":{"166435fb4feea5f4d9bfac59e79d0c8ffeec6291":{"kind":"REWORK","_number":1,"created":"2020-05-07 13:29:17.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/17/726117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/17/726117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/17/726117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/17/726117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/17/726117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/17/726117/1"}}},"commit":{"parents":[{"commit":"b9c303c2f47b4cf38393372bb0310fbf87147b50","subject":"Fix parsing of mime-type at subunit-v2-parser","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b9c303c2f47b4cf38393372bb0310fbf87147b50"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-05-07 12:29:17.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-05-07 12:29:44.000000000","tz":180},"subject":"Do not inialize scenario cls at scenario argument validator","message":"Do not inialize scenario cls at scenario argument validator\n\nChange-Id: I516d2bb46091a607122036d46ea03e7ae3ba0a67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/166435fb4feea5f4d9bfac59e79d0c8ffeec6291"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/166435fb4feea5f4d9bfac59e79d0c8ffeec6291"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
