)]}'
{"id":"openstack%2Frally~726118","triplet_id":"openstack%2Frally~master~I23b7231ccf2f9e6982612fc7733375190ab7fc98","project":"openstack/rally","branch":"master","topic":"validation","hashtags":[],"change_id":"I23b7231ccf2f9e6982612fc7733375190ab7fc98","subject":"Improve `restricted_parameters` validator to cover None case","status":"MERGED","created":"2020-05-07 13:29:17.000000000","updated":"2020-05-07 14:19:20.000000000","submitted":"2020-05-07 14:19:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726118-1588861160810-a92907d4","meta_rev_id":"36a90e05a97e69f41365b24ddbf7f3a4da11f316","_number":726118,"virtual_id_number":726118,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 14:19:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-05-07 13:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-05-07 13:31:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 14:19:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b9ec016cb678073f32cd5b791ae3cc67f27e9d9","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-05-07 13:29:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e209da550319bf7721a22649e83c03883023b83","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-05-07 13:31:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ec1d37090c83a421227c11cdc82f8b3ed1c7a25c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:46:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56511c958d784f84a33e25703bd82c08 : SUCCESS in 6m 50s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/7ba4f57b40bb48308657fb0a02424332 : SUCCESS in 8m 02s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/bdb45be3958e4729ad6b8e204f7b6335 : SUCCESS in 4m 14s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/77ad21698d564385a5a63af1c715ae18 : SUCCESS in 7m 35s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/250c11a28f6a4d5cbd823466ee627889 : SUCCESS in 6m 02s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/48ef642fcf56454e8bf7b847ac5b19e1 : SUCCESS in 4m 59s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/4bf114e58e8149c0a501f787ea32463a : SUCCESS in 5m 51s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/7b2461a8c91a478cb725864808fe2c0d : SUCCESS in 4m 36s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/dc4bc89fd39748c4b2ab099a9c5d599b : SUCCESS in 8m 57s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ba811a6d8b1f4483a1abec9d695d1716 : SUCCESS in 7m 38s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/306e19d0e6524f48b1bcf1a302132b6b : SUCCESS in 7m 40s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/750e491587104d728099e032a92ba09f : SUCCESS in 7m 59s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/2a11519900c34c659c9f9aabab1a81d6 : SUCCESS in 7m 55s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/0377caec5bf44a479eb448d3b76dcade : SUCCESS in 7m 25s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/248d018e277b454ab1b9e5d54c0decea : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1d5bbf0519a97a655ddfdb2053be745a10add5b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:51:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c8e05c3cead8c63207ff1c8c47a2cfc44096f69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:19:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"da469fc89b548b7fca0361573287d06ee1a437e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:19:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7290d3b04a3a40fe8e5b062f510e5c40 : SUCCESS in 7m 03s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/5ece47cb44ea4efdb8868000f742f7ba : SUCCESS in 8m 05s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/58a045b2450f493eaa50f555b6f62abf : SUCCESS in 4m 38s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4b110c984394abdb2029525d5693a93 : SUCCESS in 5m 22s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/9b45c3e772ca4cd49542226be9f49238 : SUCCESS in 5m 08s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/80aa82cbe14e4f3fa27d493740becf1d : SUCCESS in 8m 23s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/bcf11e77910a42159fcef440f6ec2f8f : SUCCESS in 6m 19s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ce20644fafd54f37a83b60bf9d585131 : SUCCESS in 9m 09s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/f00da6a5e6844b72a537667b65ebe974 : SUCCESS in 11m 05s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/320490e5838b4c6b94df6aaa54f4f44e : SUCCESS in 8m 54s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/5f5775fc1b394d4fb9397e930875748b : SUCCESS in 12m 22s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/53bd125dd76949a1b94f5be57e0ffd35 : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9708b288dfbc6dc5db3096209c489f6fd1c1d5e3","revisions":{"9708b288dfbc6dc5db3096209c489f6fd1c1d5e3":{"kind":"REWORK","_number":1,"created":"2020-05-07 13:29:17.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/18/726118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/18/726118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/726118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/726118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/726118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/18/726118/1"}}},"commit":{"parents":[{"commit":"166435fb4feea5f4d9bfac59e79d0c8ffeec6291","subject":"Do not inialize scenario cls at scenario argument validator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/166435fb4feea5f4d9bfac59e79d0c8ffeec6291"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-05-07 12:50:23.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-05-07 13:20:36.000000000","tz":180},"subject":"Improve `restricted_parameters` validator to cover None case","message":"Improve `restricted_parameters` validator to cover None case\n\nValidator raised the following error in case of None as target\n\n  AttributeError: \u0027NoneType\u0027 object has no attribute \u0027get\u0027\n\nChange-Id: I23b7231ccf2f9e6982612fc7733375190ab7fc98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9708b288dfbc6dc5db3096209c489f6fd1c1d5e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9708b288dfbc6dc5db3096209c489f6fd1c1d5e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
