)]}'
{"id":"openstack%2Frally~726188","triplet_id":"openstack%2Frally~master~Ie52450f1ea2f8518f4d9e910ec990752b38478f9","project":"openstack/rally","branch":"master","hashtags":[],"change_id":"Ie52450f1ea2f8518f4d9e910ec990752b38478f9","subject":"Propose Rally 3.1.0","status":"MERGED","created":"2020-05-07 15:58:31.000000000","updated":"2020-05-08 11:47:47.000000000","submitted":"2020-05-08 11:47:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726188-1588938467889-c134243d","meta_rev_id":"082f5fc940b1da3770e262491b1dc0a0309bb5e3","_number":726188,"virtual_id_number":726188,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 11:47:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-05-08 11:13:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-05-08 11:13:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-08 11:47:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5508621e377e2ef8fcfd4a28de74fae42dc47619","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-05-07 15:58:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccb84a5fa2f9c8efab6507e4075204b4de417a0c","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-05-07 16:02:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7b9ed471e3f0d932bc3e613e3a69bf68e5b2c9f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 16:21:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81003ba6bc1480a97ea5dfa06ae5167 : SUCCESS in 7m 26s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/524f40d46ccd41049a8b3b75bd1239e8 : SUCCESS in 9m 34s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/b631d28b54b04003afec59896ff1fe0d : SUCCESS in 6m 11s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/920e08aaedce4b7094a9b62955ef052d : SUCCESS in 9m 04s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/43da0b936f2c4d02b487b97567121347 : SUCCESS in 6m 38s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/e995c934f052488d8304ab3c9f346ff7 : SUCCESS in 5m 37s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/963df4c246a34abba62440ddb698c63f : SUCCESS in 5m 27s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/5b91365275ae4af593541adc86a038bd : SUCCESS in 8m 55s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/113ac75a00d2473d87286d4d1199da6f : SUCCESS in 9m 12s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b8a8599174d641b6a6dbaf184f4e84fa : SUCCESS in 9m 07s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/479c213c5c34429a98f17200a28413ac : SUCCESS in 9m 15s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/282242d3bf834497b029c4de82ced387 : SUCCESS in 7m 42s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/93d9c604f2bf48d2b60c15a229887ab0 : SUCCESS in 9m 46s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/8ae9dea217ac437dad98a5788fa715d3 : SUCCESS in 9m 02s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/a3be8abb7994425b82544a86579c7462 : SUCCESS in 12m 38s","accounts_in_message":[],"_revision_number":2},{"id":"447899292e88a3e823b88550a2f2cffc304761cc","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-05-08 11:12:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5134ba07f221f0606e8d11b4b86549dba946fef4","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-05-08 11:13:13.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"2905108a2a2a3ac256d82054f2465b324f0695a1","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-05-08 11:13:23.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"83a9ed6d6ddd140eceadaa3c7f0d0410f115cda1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 11:31:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7110d73913341f0b484e017bbaf37cc : SUCCESS in 5m 03s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/91fa82820ff04e99ae7702a91dc1a6f5 : SUCCESS in 8m 07s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/34418ef18be94479bb03cd2afbf22250 : SUCCESS in 6m 08s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/e54faa187cd5404eb6e25f1485cfae13 : SUCCESS in 4m 15s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/58389c0a0150448898752b4c9dcfa48d : SUCCESS in 6m 52s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/78567c6483df46098fafec8f8ea373e5 : SUCCESS in 7m 11s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/5a8badd655c34232b51a74d8406d9ad4 : SUCCESS in 7m 07s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/dcabdf91dc96493093881cc9b55fe659 : SUCCESS in 4m 04s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/6d3af00297ca4c8db031b94baa0e6314 : SUCCESS in 7m 44s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1cf3918cadaa49ff8d71288215386e1b : SUCCESS in 11m 36s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/2fca10304fb94e91bfdb38014932b7d8 : SUCCESS in 8m 04s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba4069cd432644b19b1e9b44ccac7e37 : SUCCESS in 8m 09s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/b8b0fc9e57c145dc8beb58146c9ca4b2 : SUCCESS in 8m 51s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/6cd8f1676dbb4aa8ad7f5debba5c99d9 : SUCCESS in 8m 45s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/72aba76f3300499497b98cae9006d600 : SUCCESS in 9m 53s","accounts_in_message":[],"_revision_number":4},{"id":"ceaa720ac5b39a1bc3bc91402831cb9a2971cbda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 11:32:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5cb0f2fca982f94348c930132040cf4710030377","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 11:47:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b49b658ca8ee44539eef465673e09306 : SUCCESS in 7m 59s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/31a942d983244365a8b5336f6881deaf : SUCCESS in 8m 13s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/d268adaa451343bfadab45218b4134a1 : SUCCESS in 7m 21s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/d09a32c25ade4309b992e67b76e9e471 : SUCCESS in 4m 59s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/93b89fdf67ee481e929c0e20ee3f2aad : SUCCESS in 3m 50s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/59d4cb5315b04b58a3ac76f297faf993 : SUCCESS in 5m 46s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/ba07ea9764984f93a8036a07b600fcb7 : SUCCESS in 8m 03s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/a1548906b9924b59838e368d53e07c1d : SUCCESS in 8m 29s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/c5c626e24b9044588431f6812a4d09d0 : SUCCESS in 8m 43s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4269e5eff1964bc79ea408ea06582c06 : SUCCESS in 7m 41s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/b9c8663b28434aedbd9c1ccb906ba229 : SUCCESS in 10m 52s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/9c4b9ace7ba84be4b57ed832758d09d7 : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":4},{"id":"5ff02aba8e0ebc0fd08276509f41aad9e2709f2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 11:47:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9beca25f0a4b05635c2eaddfaf6d8c06e037751d","revisions":{"867308293fabfa79b5b234de94da9e486cb5d8ed":{"kind":"REWORK","_number":1,"created":"2020-05-07 15:58:31.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/88/726188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/88/726188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/88/726188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/88/726188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/88/726188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/88/726188/1"}}},"commit":{"parents":[{"commit":"266d6bc53f28ae8d10013e02eacbace02f576c6b","subject":"Merge \"Improve `restricted_parameters` validator to cover None case\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/266d6bc53f28ae8d10013e02eacbace02f576c6b"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-05-07 15:58:26.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-05-07 15:58:26.000000000","tz":180},"subject":"Propose Rally 3.1.0","message":"Propose Rally 3.1.0\n\nChange-Id: Ie52450f1ea2f8518f4d9e910ec990752b38478f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/867308293fabfa79b5b234de94da9e486cb5d8ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/867308293fabfa79b5b234de94da9e486cb5d8ed"}]},"branch":"refs/heads/master"},"cc81089a6b461d8384361d7ad0a011a18fbd9f92":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-07 16:02:06.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/88/726188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/88/726188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/88/726188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/88/726188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/88/726188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/88/726188/2"}}},"commit":{"parents":[{"commit":"346268901e286b300c4ab0da08152f4979d5c670","subject":"[docs] Include rally-openstack plugins into docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/346268901e286b300c4ab0da08152f4979d5c670"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-05-07 15:58:26.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-05-07 16:02:06.000000000","tz":0},"subject":"Propose Rally 3.1.0","message":"Propose Rally 3.1.0\n\nChange-Id: Ie52450f1ea2f8518f4d9e910ec990752b38478f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cc81089a6b461d8384361d7ad0a011a18fbd9f92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cc81089a6b461d8384361d7ad0a011a18fbd9f92"}]},"branch":"refs/heads/master"},"b51b2f2eff8fc3710860b2503591d8784e682620":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-08 11:12:59.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/88/726188/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/88/726188/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/88/726188/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/88/726188/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/88/726188/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/88/726188/3"}}},"commit":{"parents":[{"commit":"88b818f2224a52e8083ffacaf614493b33366258","subject":"Update rally config samples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/88b818f2224a52e8083ffacaf614493b33366258"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-05-07 15:58:26.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-05-08 11:12:59.000000000","tz":0},"subject":"Propose Rally 3.1.0","message":"Propose Rally 3.1.0\n\nChange-Id: Ie52450f1ea2f8518f4d9e910ec990752b38478f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b51b2f2eff8fc3710860b2503591d8784e682620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b51b2f2eff8fc3710860b2503591d8784e682620"}]},"branch":"refs/heads/master"},"9beca25f0a4b05635c2eaddfaf6d8c06e037751d":{"kind":"REWORK","_number":4,"created":"2020-05-08 11:13:13.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/88/726188/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/88/726188/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/88/726188/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/88/726188/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/88/726188/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/88/726188/4"}}},"commit":{"parents":[{"commit":"88b818f2224a52e8083ffacaf614493b33366258","subject":"Update rally config samples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/88b818f2224a52e8083ffacaf614493b33366258"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-05-07 15:58:26.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-05-08 11:13:09.000000000","tz":0},"subject":"Propose Rally 3.1.0","message":"Propose Rally 3.1.0\n\nChange-Id: Ie52450f1ea2f8518f4d9e910ec990752b38478f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9beca25f0a4b05635c2eaddfaf6d8c06e037751d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9beca25f0a4b05635c2eaddfaf6d8c06e037751d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
