)]}'
{"id":"openstack%2Frally~726341","triplet_id":"openstack%2Frally~master~Id17e9a62a6f0302004522e7b78c5d5aac154e59c","project":"openstack/rally","branch":"master","topic":"conf_sample","hashtags":[],"change_id":"Id17e9a62a6f0302004522e7b78c5d5aac154e59c","subject":"Update rally config samples","status":"MERGED","created":"2020-05-08 11:12:42.000000000","updated":"2020-05-08 11:44:48.000000000","submitted":"2020-05-08 11:44:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726341-1588938288004-39316be2","meta_rev_id":"8346ae9bd22132252ae3577cfaa29ec34565a4c5","_number":726341,"virtual_id_number":726341,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 11:44:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-05-08 11:13:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-05-08 11:13:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-08 11:44:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2ee8a1eeeda73141fb6b66a69199c67cd3493cb","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-05-08 11:12:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94393b2b2eae12380ed026490fab5c9d17bd9722","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-05-08 11:13:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d12f33f8abe8131046b9c2e3765c56e87813c86d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 11:28:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34898fdf2f2447d1b5b6d4ed40b2e867 : SUCCESS in 5m 12s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/f86a458f21264d29a94738ae3b96241d : SUCCESS in 6m 08s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/cd28457cfe32437abd92ccaf461b91aa : SUCCESS in 5m 29s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/35a6c9e454f84847b7c7ef63cf97568a : SUCCESS in 5m 42s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/997f3665cbe7459a97178278e243e8ca : SUCCESS in 4m 24s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/67564871f2a04be58c9c89956cd1ce77 : SUCCESS in 5m 03s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/269e85d5bba844f2a7eed1adb0636250 : SUCCESS in 5m 54s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/ec2ee90906ea414c8c6dada3ee7230d2 : SUCCESS in 4m 10s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/4d4486608d574a18a69e085eb6673ab0 : SUCCESS in 7m 24s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4f83b869e81b47b3b06013c81d7f6fa4 : SUCCESS in 11m 19s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/9c6c908995fd44328474d685964f1630 : SUCCESS in 9m 30s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3d01c0a790e41838cb3a3a497613053 : SUCCESS in 8m 11s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/f48a9b0c45ed4703b000c6462e650150 : SUCCESS in 8m 59s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/6e701c10958842e483ae057817e99a77 : SUCCESS in 8m 04s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/f7c0a69ac9ef48f4b7ce8cb442c88f75 : SUCCESS in 9m 40s","accounts_in_message":[],"_revision_number":1},{"id":"69658e92c347c9912432bb831ec36517b759b9e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 11:29:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a9a1c2da6720e36d1c06385acc6ea30198f96ceb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 11:44:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5283a315dee74636b777ce6b751f1318 : SUCCESS in 7m 21s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/2c4d4f1daca14ea79a690869f2a5fea1 : SUCCESS in 8m 06s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/7c5832befc2d4241bcacd4d845c669cb : SUCCESS in 5m 01s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/f545f707a7644161bf63ad1bf0f22c4e : SUCCESS in 5m 44s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/8ff1fdcd32394c89b74559f6571bb5e9 : SUCCESS in 5m 32s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/ed080c9d98be44b4a62fdd05f2985c2c : SUCCESS in 7m 56s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/75165649650049e38d17929bad0423ca : SUCCESS in 5m 41s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/e1e737ec445a416db2b830598159b8f5 : SUCCESS in 9m 18s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/c51be0ab14a040ec89dc1aaf6e3659f6 : SUCCESS in 9m 03s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afc3d62fbeee4bb28e1a74faf46fc249 : SUCCESS in 7m 03s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/62160b6cb77b4bbca3587d2cb127a8a2 : SUCCESS in 8m 26s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/92df15668810423786aec75211358169 : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b58dad4f80750b2ce500de22bfe8d9f882f17274","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 11:44:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88b818f2224a52e8083ffacaf614493b33366258","revisions":{"88b818f2224a52e8083ffacaf614493b33366258":{"kind":"REWORK","_number":1,"created":"2020-05-08 11:12:42.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/41/726341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/41/726341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/726341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/726341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/726341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/41/726341/1"}}},"commit":{"parents":[{"commit":"266d6bc53f28ae8d10013e02eacbace02f576c6b","subject":"Merge \"Improve `restricted_parameters` validator to cover None case\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/266d6bc53f28ae8d10013e02eacbace02f576c6b"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-05-08 11:11:37.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-05-08 11:11:37.000000000","tz":180},"subject":"Update rally config samples","message":"Update rally config samples\n\n+ include 2 important config options:\n\n   * scenario_resource_name_format\n   * and context_resource_name_format.\n\nChange-Id: Id17e9a62a6f0302004522e7b78c5d5aac154e59c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/88b818f2224a52e8083ffacaf614493b33366258"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/88b818f2224a52e8083ffacaf614493b33366258"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
