)]}'
{"id":"openstack%2Frally~728733","triplet_id":"openstack%2Frally~master~I94ae0a77aa8c96b842e62822c54c326192d75b55","project":"openstack/rally","branch":"master","topic":"hacking-fix","hashtags":[],"change_id":"I94ae0a77aa8c96b842e62822c54c326192d75b55","subject":"Fix hacking min version to 3.0.1","status":"ABANDONED","created":"2020-05-18 01:17:50.000000000","updated":"2020-05-19 00:22:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"63a7715ea9586ed6d96dcbca751192bf03c0a73a","_number":728733,"virtual_id_number":728733,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"date":"2020-05-18 09:46:43.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-18 01:46:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":-2,"date":"2020-05-18 08:20:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-18 01:46:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-18 08:20:06.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"0c6f1b392bb2299db2164b3f8669625bfefd6697","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-05-18 01:17:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7728007d95c9ebf41a8f4e175e29f6d454ded993","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 01:30:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa1fcc2f78e1428ebeddd42ecc3244a4 : SUCCESS in 4m 40s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/a4199d378c9b43af8c7fb99055e81490 : SUCCESS in 5m 25s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/21a3a10030eb4bb2840a342680a69546 : SUCCESS in 4m 44s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ab0fdc883a84970a0a840f1e5f51747 : SUCCESS in 6m 03s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/898727960def4ed784700f01fe0615e7 : SUCCESS in 4m 11s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/773bb6d5b6724223850327feb7ffedc3 : SUCCESS in 4m 01s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/b6bee7f3c5a145438d52bdb768752618 : SUCCESS in 4m 12s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/9def6db61dea44d18466a9cb559f2ee2 : SUCCESS in 4m 24s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/06089e9911344f31b23c3aedc1363d80 : SUCCESS in 7m 00s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/039b0dd691fa4f11b7ba0f4ab3dc9904 : SUCCESS in 9m 20s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/021b2a081ba5487c880fe80abf74cc3c : SUCCESS in 7m 46s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3546f39d15fe404db92473816f021043 : SUCCESS in 6m 15s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/d0b32abb42094cd9b24dc8736c46e10d : SUCCESS in 7m 45s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/b334dedf43b149da80c8231046f022f4 : SUCCESS in 6m 33s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/6c8816e6a0fe4ed1a6f775f03c389dd3 : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":1},{"id":"bd696cb5cf4c41623a7a897750a966e0427229ab","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-05-18 01:35:40.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"dc16a97434f306516ede92bb29ac52a614d65649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 01:46:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/033320366dab45cfaff5431e7277cde7 : SUCCESS in 5m 18s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/c34810d9a2b446c4bbc1cdf8900eafd3 : SUCCESS in 5m 08s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/77edaaf3b7034ae3b5d9c85fdd7ca869 : SUCCESS in 4m 39s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a1502757e9643cf9da2ea7da1ad8daa : SUCCESS in 3m 23s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/5d286151efd74caba9c55fbee9264ff9 : SUCCESS in 4m 05s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/588b4c7096044d7383f0962383804638 : SUCCESS in 4m 38s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/62a195c3ca1d41b2bd0fe9613b0c6135 : SUCCESS in 4m 52s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/441bde7083124f859c2dc7676d4ae840 : SUCCESS in 3m 44s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/dd29967212e44be8badc460043b74c4f : SUCCESS in 6m 58s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/81f739a26bd34a278bdd99b47bac8f9e : SUCCESS in 8m 51s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/eedd7b61c0264d04935d14186317c0d6 : SUCCESS in 7m 12s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac74b99fd6a1461bafb0072ddd7b8736 : SUCCESS in 6m 29s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/9e891be161ba4e05bec9a1cde4892753 : SUCCESS in 7m 50s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/ab8bb6fe6b874b53a338f1af36be55a9 : SUCCESS in 7m 16s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/fe6d09505f634c4fb794d10e0cb097ca : SUCCESS in 9m 28s","accounts_in_message":[],"_revision_number":2},{"id":"8274dcd62cfee6f426ed137cfec37cd5f70e9922","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-05-18 08:20:06.000000000","message":"Patch Set 2: Code-Review-2\n\nThanks for your work, but we already fixed new flake8 failures","accounts_in_message":[],"_revision_number":2},{"id":"4bf9e215601cc6af17954ffeb74af26630269d85","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-05-18 09:34:17.000000000","message":"Patch Set 2:\n\n\u003e Thanks for your work, but we already fixed new flake8 failures\n\nYou mean https://review.opendev.org/#/c/727136/1 this patch? I dont find any patch to bump hacking min version to 3.0.1, or I missed something?","accounts_in_message":[],"_revision_number":2},{"id":"826692ee27eddcf9b5054339ec19a58853d5a68e","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-05-18 09:46:43.000000000","message":"Patch Set 2:\n\n\u003e Thanks for your work, but we already fixed new flake8 failures\n\nThis patch maily to do \"To avoid similar gate break in future, we need to bump the hacking min version.\"\n\nI think we need.","accounts_in_message":[],"_revision_number":2},{"id":"3492a438e1ca400f7d25a788e6f59156c0374156","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-05-18 12:24:09.000000000","message":"Patch Set 2:\n\n\u003e You mean https://review.opendev.org/#/c/727136/1 this patch?\n\nYes, this patch includes all fixes.\n\n\u003e I dont find any patch to bump hacking min version to 3.0.1, or I missed something?\n\nI made a fix for a new flake 8 release before hacking\u003d\u003d3.0.1 was introduced. \n\n\u003e This patch maily to do \"To avoid similar gate break in future, we need to bump the hacking min version.\"\n\nI\u0027m ok to bump min version of hacking, but against to pin it to a specific version.","accounts_in_message":[],"_revision_number":2},{"id":"90a06f5aada64b1535cc1d13409c973cd11dd2ff","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-05-19 00:22:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f53e48717185f6c64dea971aaae14ec7a5725c85","revisions":{"fa2b98629adee043fae0d1d2a639de901183f946":{"kind":"REWORK","_number":1,"created":"2020-05-18 01:17:50.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/33/728733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/33/728733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/33/728733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/33/728733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/33/728733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/33/728733/1"}}},"commit":{"parents":[{"commit":"e1ad05c781fdabd013d81277b459caf6db1a4279","subject":"Make config parser case sensitivity in configure-verifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e1ad05c781fdabd013d81277b459caf6db1a4279"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-05-18 01:17:01.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-05-18 01:17:42.000000000","tz":480},"subject":"Bump hacking min version to 3.0.1","message":"Bump hacking min version to 3.0.1\n\nhacking 3.0.1 fix the pinning of flake8 to avoid bringing in a new\nversion with new checks.\n\nbumping the min version for hacking so that any older hacking versions\nwhich auto adopt the new checks are not used.\n\nChange-Id: I94ae0a77aa8c96b842e62822c54c326192d75b55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/fa2b98629adee043fae0d1d2a639de901183f946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/fa2b98629adee043fae0d1d2a639de901183f946"}]},"branch":"refs/heads/master"},"f53e48717185f6c64dea971aaae14ec7a5725c85":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-18 01:35:40.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/33/728733/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/33/728733/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/33/728733/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/33/728733/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/33/728733/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/33/728733/2"}}},"commit":{"parents":[{"commit":"e1ad05c781fdabd013d81277b459caf6db1a4279","subject":"Make config parser case sensitivity in configure-verifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e1ad05c781fdabd013d81277b459caf6db1a4279"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-05-18 01:17:01.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-05-18 01:35:35.000000000","tz":0},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nflake8 new release 3.8.0 added new checks and gate pep8\njob start failing. hacking 3.0.1 fix the pinning of flake8 to\navoid bringing in a new version with new checks.\n\nThough it is fixed in latest hacking but 2.0 and 3.0 has cap for\nflake8 as \u003c4.0.0 which mean flake8 new version 3.9.0 can also\nbreak the pep8 job if new check are added.\n\nTo avoid similar gate break in future, we need to bump the hacking min\nversion.\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014828.html\n\nChange-Id: I94ae0a77aa8c96b842e62822c54c326192d75b55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f53e48717185f6c64dea971aaae14ec7a5725c85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f53e48717185f6c64dea971aaae14ec7a5725c85"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
