)]}'
{"id":"openstack%2Frally~737461","triplet_id":"openstack%2Frally~master~Ifc0f70adf3a426314ff6df41c58b6aa5aeca17fb","project":"openstack/rally","branch":"master","topic":"sphinx","hashtags":[],"change_id":"Ifc0f70adf3a426314ff6df41c58b6aa5aeca17fb","subject":"Fix pygments style","status":"MERGED","created":"2020-06-23 09:05:18.000000000","updated":"2020-07-02 12:56:11.000000000","submitted":"2020-07-02 12:56:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737461-1593694571226-f3086ce0","meta_rev_id":"a95e489a2c57ddb12cf7c0ba39167ca3cfbb48a8","_number":737461,"virtual_id_number":737461,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-02 12:56:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-07-02 12:16:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-07-02 12:16:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-02 12:16:24.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2020-07-02 12:56:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9980d193b87f93f757600a666e73921d91726501","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-06-23 09:05:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c6cec08dbd93e4255098dec4d2b552ad2a787d8","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-06-23 09:20:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a298e64c2de5c917e93176b2d15893dc0d34ee9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 09:29:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e609f5886f4ec59aff6f4dc313fc39 : SUCCESS in 5m 56s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/7c96aab2feff45728c748ed06b25d7a0 : SUCCESS in 6m 20s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/a9fe7f0a60fc4e1bb1c684ac484208dc : SUCCESS in 8m 24s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/289dde6224fa4d838461bd7a9d0396f3 : SUCCESS in 5m 37s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/d114408419b34c2cb84171022ecdb85a : SUCCESS in 5m 38s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/6130b3e36ba24721b1c7737ed50fe94b : SUCCESS in 4m 33s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/1d8500ea1744400d8fb919ccab827ab4 : SUCCESS in 6m 45s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/a3e29c23a8e242a89fe9c3610558e6ec : SUCCESS in 3m 41s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/dc382a3fd40e4687b8a1ac8206d96d8b : SUCCESS in 6m 34s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3f639e4e5ab74365a5258396f1b60eee : SUCCESS in 8m 16s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/bc1b7f5120164f6d890220ac2d77b9d5 : SUCCESS in 9m 57s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d049d73caf84e2daab31df18e3a56d2 : RETRY_LIMIT in 3m 16s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/58eee27ac41043f5a4d93e64ef14f7bf : SUCCESS in 11m 00s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/cb987c1d8e09406cb21e77b3a4ce7610 : SUCCESS in 7m 09s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/b2bcc275ceeb4c09beaff8460ed0ef18 : SUCCESS in 12m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b1b7908c007284934d727ef4d7c2fe221a5a3dec","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-06-24 16:14:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a82c42cf6ab2bcf2370856de975fe9d9dfb1f3cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 16:35:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2efedfa37e8747b6a3280c3f97c9b402 : SUCCESS in 7m 03s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/519cdd1b5adb410b90131a5cbf304ec3 : SUCCESS in 7m 57s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ca03677a018d4d329f98007f4ab88036 : SUCCESS in 7m 09s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/7698ba2b58174e699ba5611b343c897b : SUCCESS in 7m 07s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/9e6404201e0b4bcaa4100f4f3c678059 : SUCCESS in 4m 51s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/6c7e79f381e446e6958d357c12ca4e28 : SUCCESS in 6m 20s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/617054807ff24f8db5b49e8503205f39 : SUCCESS in 7m 16s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/f518b055492c41498f1b99c4fa9e06d6 : SUCCESS in 5m 40s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/16e0d1b49626427288244d8e46ad63a1 : SUCCESS in 11m 13s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bed7975cb12648e7933070b0cc4d6459 : SUCCESS in 9m 25s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/44091bbed1b24fc8aebc9b19cb8cb536 : SUCCESS in 9m 45s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03d3093d2c094654ac633142e91fdb12 : RETRY_LIMIT in 2m 03s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/0729c18d3acd46dd99873b4c66a232d2 : SUCCESS in 15m 10s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/04539a94f44a4c9e8e5209293c2b0dad : SUCCESS in 7m 53s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/55013d3c18bc401991a668d3467eddfb : SUCCESS in 10m 11s","accounts_in_message":[],"_revision_number":1},{"id":"9bdd8f41e5c826fe27a944a98229b51d6ac9abc8","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-07-02 06:25:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b33b20b8443e56bb1a657eb48b0f3405df2d123c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 06:54:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42ff684fae7840679e100491ca0a8bc4 : SUCCESS in 7m 06s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/446a857c5f6648fc8774c436475f10aa : SUCCESS in 9m 41s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f3deea86c4614f2d8a11cd4b70a54f09 : SUCCESS in 9m 33s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6335d7ea9e4435da05fa1ccfa6f7223 : SUCCESS in 7m 00s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/c12864cd166e41fb92181a6f64fb540f : SUCCESS in 8m 49s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/060aaea087b74a16948e9f39762428f9 : SUCCESS in 7m 05s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/15d79f6033f848cc8c62977614296f8e : SUCCESS in 6m 59s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/ccabd6fcac7f4e6484d4908db2fcfb94 : SUCCESS in 6m 01s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/51e476177bd5492497e59099f7f86d5a : SUCCESS in 9m 10s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/32456439b21342f292e5f538a148e1d9 : SUCCESS in 12m 02s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/b124e1a2a1504c70841253d5b3bba36c : SUCCESS in 8m 51s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bcf8412a03a04821b58732c78da9a066 : RETRY_LIMIT in 3m 56s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/ae2cd3123ba74aa58e1af287f61fbc05 : SUCCESS in 12m 12s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/942a0688181845f2851b5574376e44cc : SUCCESS in 9m 50s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/139d6ac1eebd4735a76ae4ecc9348d86 : SUCCESS in 11m 04s","accounts_in_message":[],"_revision_number":1},{"id":"f238ca50fc7d73331d1883df795d7ba9aab3dc3b","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-07-02 12:16:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"869ecd14f444f9f6944039f406e0fdb8bbafd0d0","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-07-02 12:16:24.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a834ac8dc1625a3678e59e8abd48ff10ffbdbc2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:40:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82fe57e9901648a0a0f027e827ad86d4 : SUCCESS in 10m 33s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e0f23443c9a14ba6abc059e2add0239b : SUCCESS in 7m 30s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/5af3abc5088f493ba1f8bc85844e66f4 : SUCCESS in 6m 31s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ce26b1d11743b3a8d479869da1cf96 : SUCCESS in 4m 26s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/2c1cdcb7b72c4b3ab94146bef38e9a9d : SUCCESS in 6m 26s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/b57c78569e7f4f0ca91d23328b8c5943 : SUCCESS in 6m 02s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/4f8c4210e60c43aeac6174c8f7f1766d : SUCCESS in 6m 30s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/c731bf7817da4154933c941060b9a4be : SUCCESS in 4m 02s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/d72679f0064a47d18b95243bfd439162 : SUCCESS in 9m 05s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0dcbf154a813435a99e3daa7ba52c305 : SUCCESS in 7m 14s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/0e478b9245e14d24a1b429fc545b6dc5 : SUCCESS in 8m 09s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7f226ada43746258b10ce6173724da1 : SUCCESS in 8m 16s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9a2e33be88b417381c9633fd58e6dbc : SUCCESS in 6m 33s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/32555ef2b7e24a0e8cb009b0466337e4 : SUCCESS in 7m 34s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/41a1c7b899ed4c8ca7b0cacbd17a7551 : SUCCESS in 8m 25s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/8118a5680ee540339c6898205ddd2ebe : SUCCESS in 10m 16s","accounts_in_message":[],"_revision_number":2},{"id":"777de2c90bdd9b74cfbeceb5448f58879db08a8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:40:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8b18590ad401351abf4a0c86970395eab81e6ee4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:56:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a4b755397fb4c7085e34a2f6468657e : SUCCESS in 7m 58s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b5e845666fcf4d71b117b7e475abf9ae : SUCCESS in 6m 20s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/7fd09275d4184077a6af116c7adc0e50 : SUCCESS in 7m 30s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d77ae1b0d8a4c089d174aaf0168b014 : SUCCESS in 4m 55s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/20a67947b4cb4e8d8a89418c309cabfb : SUCCESS in 6m 00s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/d3d0660658e5479791468cbc437d2676 : SUCCESS in 7m 25s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/83722fd15cc3421b8d6d1671b2d63ad3 : SUCCESS in 9m 26s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/32b52ef9f5fd46799602732605b20c1a : SUCCESS in 7m 17s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/defb2dd68713432cb94db1807411c30e : SUCCESS in 10m 02s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc7777b2b3e24af8908a9dec966b060b : SUCCESS in 9m 21s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cb656a9ff074c6ba0158e2ac861a8af : SUCCESS in 8m 17s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/9ac9b2c958754d6cb04b4e7ae41d19c8 : SUCCESS in 7m 17s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/f53c402ae4b9439eb24ec2eca6e28045 : SUCCESS in 9m 39s","accounts_in_message":[],"_revision_number":2},{"id":"cc80c0dd5482bb5b7b5d36eb38b4f734faaf57ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:56:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b268d60119113ed89107815b50dc496aaea9af8","revisions":{"ee2c6bdd9c71502430ded45f9425327574ce99f7":{"kind":"REWORK","_number":1,"created":"2020-06-23 09:05:18.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/61/737461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/61/737461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/61/737461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/61/737461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/61/737461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/61/737461/1"}}},"commit":{"parents":[{"commit":"e1ad05c781fdabd013d81277b459caf6db1a4279","subject":"Make config parser case sensitivity in configure-verifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e1ad05c781fdabd013d81277b459caf6db1a4279"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-06-23 09:03:36.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-06-23 09:03:36.000000000","tz":480},"subject":"Fix pygments style","message":"Fix pygments style\n\nNew theme of docs respects pygments_style.\n\nmore info: http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: Ifc0f70adf3a426314ff6df41c58b6aa5aeca17fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ee2c6bdd9c71502430ded45f9425327574ce99f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ee2c6bdd9c71502430ded45f9425327574ce99f7"}]},"branch":"refs/heads/master"},"7b268d60119113ed89107815b50dc496aaea9af8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-02 12:16:19.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/61/737461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/61/737461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/61/737461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/61/737461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/61/737461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/61/737461/2"}}},"commit":{"parents":[{"commit":"8087d7caa1386a9d3da32e1199e7374576f7c528","subject":"Add ubuntu-focal install job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8087d7caa1386a9d3da32e1199e7374576f7c528"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-06-23 09:03:36.000000000","tz":480},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-07-02 12:16:19.000000000","tz":0},"subject":"Fix pygments style","message":"Fix pygments style\n\nNew theme of docs respects pygments_style.\n\nmore info: http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: Ifc0f70adf3a426314ff6df41c58b6aa5aeca17fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7b268d60119113ed89107815b50dc496aaea9af8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7b268d60119113ed89107815b50dc496aaea9af8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
