)]}'
{"id":"openstack%2Frally~739016","triplet_id":"openstack%2Frally~master~Ic5e4f41f142d978d5e85b27ea324805bef307983","project":"openstack/rally","branch":"master","topic":"builds_fix","hashtags":[],"change_id":"Ic5e4f41f142d978d5e85b27ea324805bef307983","subject":"Add ubuntu-focal install job","status":"MERGED","created":"2020-07-02 10:46:10.000000000","updated":"2020-07-02 12:36:19.000000000","submitted":"2020-07-02 12:36:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739016-1593693379539-8c5a5b72","meta_rev_id":"d4c60cbfb6ccee3731d098b1037023d7fcd1f614","_number":739016,"virtual_id_number":739016,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-02 12:36:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-07-02 12:16:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-07-02 12:16:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-02 12:36:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52849f9fdd01cea517aaa733a4daa2e17e4ac325","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-07-02 10:46:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08bcf2b182566ef7bcc47ddd4b56fd7ad7424089","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-07-02 10:54:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0255a5620d4f2a3b9c9ff9ddb072309394be198c","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-07-02 11:15:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d90f4641c9d9b77c8cf13c7d6a423f4a7dcf3994","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-07-02 11:24:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"10badb8dfca8eccafaff6f33f25341b44935e5d9","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-07-02 11:46:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f4a9b5fd67b7e33db5ffbcd5c1a72c1c70184e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:06:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56cc950920af485baa33c9d902d12dfa : SUCCESS in 6m 16s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/fdfc51cbcee94297985e1a333ba9b76a : SUCCESS in 4m 08s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/df9e52557d224e83945295faa114f5b7 : SUCCESS in 7m 55s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/36af73144dc04ad4be3f7cb4b8744cff : SUCCESS in 5m 00s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/13c5d901173f44209ef5e6ef1c485222 : SUCCESS in 4m 53s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/ed6ab69321aa452b8522437575cf2096 : SUCCESS in 5m 52s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/f81e16349fd147dd87c68f01441c8e2f : SUCCESS in 5m 03s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/432aad19e8514b909154996302e40375 : SUCCESS in 5m 16s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/168e2686cddc4bea80b0364008449cce : SUCCESS in 7m 58s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/81a168c20a8443b39c968167ea56d0b0 : SUCCESS in 6m 13s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ccc963ea36424a7993e7a94acb417046 : SUCCESS in 10m 45s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a79aafb1e26c49eea6cf22f3eff34c5a : SUCCESS in 6m 48s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57fbfa708f794c36871d7ce4fb615fa6 : SUCCESS in 7m 01s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/e0c65415363d4e8286641ce85888bbb5 : SUCCESS in 7m 13s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/6985ee1264bf44c69b18807e692ace64 : SUCCESS in 8m 22s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/91d7d6c1da6548c2b89ee48e67e81cb5 : SUCCESS in 8m 44s","accounts_in_message":[],"_revision_number":5},{"id":"426db2cd96948dfe00f2b6416eb232e1fafc518a","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-07-02 12:16:00.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2eba2c9a4872b223448a37680f547cee99794ab3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:18:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b8c0f0af11100472d852bbc46dc2136c007b48a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:36:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"698e5061934e83aff480495eadad1fca0e340d97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:36:19.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e7d15dcedcd48c8b9c58996bbcc2901 : SUCCESS in 8m 59s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/ae868f72c6de4c06a37e4aada1bfcebe : SUCCESS in 10m 29s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/4591dbb3c49743b1b0205d9524efe758 : SUCCESS in 7m 26s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/10864bd7024043a3ae2be5f94fbfae7f : SUCCESS in 6m 53s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/558c3c2eece64061a89da589c60f0aba : SUCCESS in 7m 18s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/45f4bf2385fc4602ad7d04597a42c8f8 : SUCCESS in 4m 40s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/4a0276803bda4026b631cf3c855dd248 : SUCCESS in 6m 56s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/371a5fb4f93142588dfca3959118442c : SUCCESS in 9m 30s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/caf0111aa3474b3998764d57640cebb6 : SUCCESS in 9m 01s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b69d28d07f644d8b73410d18582b6f3 : SUCCESS in 6m 54s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a98547cafffe43c7beb1510de34985c5 : SUCCESS in 7m 56s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/6a0ce9434ca54e189677c05bf418bd0a : SUCCESS in 9m 31s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/a2670a32ecf74d76a9647f7df7b09824 : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8087d7caa1386a9d3da32e1199e7374576f7c528","revisions":{"970a4c19404acc48a402cebf2aec3efa8d6d603b":{"kind":"REWORK","_number":1,"created":"2020-07-02 10:46:10.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/16/739016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/16/739016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/16/739016/1"}}},"commit":{"parents":[{"commit":"e1ad05c781fdabd013d81277b459caf6db1a4279","subject":"Make config parser case sensitivity in configure-verifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e1ad05c781fdabd013d81277b459caf6db1a4279"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-07-02 10:45:51.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-07-02 10:46:04.000000000","tz":180},"subject":"Add ubuntu-focal install job","message":"Add ubuntu-focal install job\n\nChange-Id: Ic5e4f41f142d978d5e85b27ea324805bef307983\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/970a4c19404acc48a402cebf2aec3efa8d6d603b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/970a4c19404acc48a402cebf2aec3efa8d6d603b"}]},"branch":"refs/heads/master"},"855aca3ff34f49aebe4e05e7834b1dc6965ff54e":{"kind":"REWORK","_number":2,"created":"2020-07-02 10:54:27.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/16/739016/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/16/739016/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/16/739016/2"}}},"commit":{"parents":[{"commit":"e1ad05c781fdabd013d81277b459caf6db1a4279","subject":"Make config parser case sensitivity in configure-verifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e1ad05c781fdabd013d81277b459caf6db1a4279"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-07-02 10:45:51.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-07-02 10:54:21.000000000","tz":180},"subject":"Add ubuntu-focal install job","message":"Add ubuntu-focal install job\n\nChange-Id: Ic5e4f41f142d978d5e85b27ea324805bef307983\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/855aca3ff34f49aebe4e05e7834b1dc6965ff54e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/855aca3ff34f49aebe4e05e7834b1dc6965ff54e"}]},"branch":"refs/heads/master"},"516ebd713acab061bc0384d79b400f3200078551":{"kind":"REWORK","_number":3,"created":"2020-07-02 11:15:58.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/16/739016/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/16/739016/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/16/739016/3"}}},"commit":{"parents":[{"commit":"e1ad05c781fdabd013d81277b459caf6db1a4279","subject":"Make config parser case sensitivity in configure-verifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e1ad05c781fdabd013d81277b459caf6db1a4279"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-07-02 10:45:51.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-07-02 11:15:52.000000000","tz":180},"subject":"Add ubuntu-focal install job","message":"Add ubuntu-focal install job\n\nChange-Id: Ic5e4f41f142d978d5e85b27ea324805bef307983\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/516ebd713acab061bc0384d79b400f3200078551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/516ebd713acab061bc0384d79b400f3200078551"}]},"branch":"refs/heads/master"},"33d51e895e08c9a9ec96169ac1a172c5c7805358":{"kind":"REWORK","_number":4,"created":"2020-07-02 11:24:34.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/16/739016/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/16/739016/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/16/739016/4"}}},"commit":{"parents":[{"commit":"e1ad05c781fdabd013d81277b459caf6db1a4279","subject":"Make config parser case sensitivity in configure-verifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e1ad05c781fdabd013d81277b459caf6db1a4279"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-07-02 10:45:51.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-07-02 11:24:28.000000000","tz":180},"subject":"Add ubuntu-focal install job","message":"Add ubuntu-focal install job\n\nChange-Id: Ic5e4f41f142d978d5e85b27ea324805bef307983\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/33d51e895e08c9a9ec96169ac1a172c5c7805358"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/33d51e895e08c9a9ec96169ac1a172c5c7805358"}]},"branch":"refs/heads/master"},"8087d7caa1386a9d3da32e1199e7374576f7c528":{"kind":"REWORK","_number":5,"created":"2020-07-02 11:46:41.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/16/739016/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/16/739016/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/16/739016/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/16/739016/5"}}},"commit":{"parents":[{"commit":"e1ad05c781fdabd013d81277b459caf6db1a4279","subject":"Make config parser case sensitivity in configure-verifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e1ad05c781fdabd013d81277b459caf6db1a4279"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-07-02 10:45:51.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-07-02 11:46:33.000000000","tz":180},"subject":"Add ubuntu-focal install job","message":"Add ubuntu-focal install job\n\nChange-Id: Ic5e4f41f142d978d5e85b27ea324805bef307983\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8087d7caa1386a9d3da32e1199e7374576f7c528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8087d7caa1386a9d3da32e1199e7374576f7c528"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
