)]}'
{"id":"openstack%2Frally~739361","triplet_id":"openstack%2Frally~master~I4a10dda03078c51380e8faad7684d2ba810f097f","project":"openstack/rally","branch":"master","hashtags":[],"change_id":"I4a10dda03078c51380e8faad7684d2ba810f097f","subject":"Switch from unittest2 compat methods to Python 3.x methods","status":"MERGED","created":"2020-07-05 12:16:34.000000000","updated":"2020-07-08 09:40:57.000000000","submitted":"2020-07-08 09:40:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739361-1594201257611-5d897941","meta_rev_id":"d6999ff0d9484543b03e3d00f4a8d0414b54aeb2","_number":739361,"virtual_id_number":739361,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-08 09:40:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-07-08 09:24:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-07-08 09:24:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-08 09:24:00.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2020-07-08 09:40:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26f0f4c917de9f9739ca3a9a8e09382d1e3d8928","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-07-05 12:16:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6269eb68efe9b98c1f5b47871d22bdc45b647fc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 12:36:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23d9d86e9ca840f0afd4df6207d26913 : SUCCESS in 5m 10s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/379efe9b329c4ecca53835bb773889f6 : SUCCESS in 5m 06s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/e97c04208dc44a24a5094b0189dbcf4e : SUCCESS in 4m 59s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/871a28140bb24737adca5c28031ec38f : SUCCESS in 4m 03s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/4a8a1398280f42b9b67cfe2091e407c9 : SUCCESS in 4m 10s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/fcacaed9515a4a97b0fc33a374ce3108 : SUCCESS in 4m 36s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/8a5f862be97042f9b38693c1f1f0e2bc : SUCCESS in 4m 19s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/3c0b6ccc97794eb5ac00f0f56379a586 : SUCCESS in 3m 58s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/7cecac624502401d86109dcef9447003 : SUCCESS in 6m 57s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ce5d6a930fc14642b281a1d3a0fb5405 : SUCCESS in 7m 21s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/46e33b2c2f98450d9de4f383f3ddc7fe : SUCCESS in 7m 12s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/194d75d4c77049c88f0169606184463f : SUCCESS in 6m 25s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6a7afbeacc74abf8cb111a61d9561a6 : SUCCESS in 6m 17s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/5490f8f6c75d475095998cce7b7da4cd : SUCCESS in 6m 45s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/4e294964775a4296bbedd90eaf72929b : SUCCESS in 7m 01s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/c6376ba703a342fa89d434874dcac79a : SUCCESS in 9m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6c4286c803efe4e02514077c016b0bacf803d648","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-07-08 09:24:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b013829f935eee247169ecfaf5036d0a7c507b11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:24:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7bd1cc72bf2d0cca90fbd20216e4ee163e69a26c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:40:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d6999ff0d9484543b03e3d00f4a8d0414b54aeb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:40:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40828a9853294e8b9349eebed7b1d80a : SUCCESS in 8m 08s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/6b1c014712e94eeeb3e7b208ca5bdabd : SUCCESS in 6m 11s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/1949c2d4872a4996965309952059cf1a : SUCCESS in 5m 47s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d60883b6238418389f21ebcad35d9be : SUCCESS in 5m 41s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/9761e62f236d47e4aa43fd9fb05e74e6 : SUCCESS in 6m 41s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/290566148e4e47e4970f6070e54fe6e9 : SUCCESS in 4m 49s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/da85345ff09247c5aee008475257c20e : SUCCESS in 5m 23s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/db1a3d5e07e44d0b9cfcbf6629d9cc4e : SUCCESS in 7m 40s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/3e15bf4111864c7f9d921805a57c041c : SUCCESS in 10m 26s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdd13d61d9ce4a90a529b41b49f1ec21 : SUCCESS in 8m 14s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ada05e7b7b4045cd8566cd71e146d41e : SUCCESS in 9m 31s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/fdec168d385a49dba7331b05d7e47bb7 : SUCCESS in 7m 35s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/7cc05a14d5444ab3a8c3db6823297d6b : SUCCESS in 8m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a5e8945f1ef70a76ba7de172c60b049e84c68f2","revisions":{"7a5e8945f1ef70a76ba7de172c60b049e84c68f2":{"kind":"REWORK","_number":1,"created":"2020-07-05 12:16:34.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/61/739361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/61/739361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/61/739361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/61/739361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/61/739361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/61/739361/1"}}},"commit":{"parents":[{"commit":"e1ad05c781fdabd013d81277b459caf6db1a4279","subject":"Make config parser case sensitivity in configure-verifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e1ad05c781fdabd013d81277b459caf6db1a4279"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-07-05 12:08:00.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-07-05 12:08:00.000000000","tz":480},"subject":"Switch from unittest2 compat methods to Python 3.x methods","message":"Switch from unittest2 compat methods to Python 3.x methods\n\nWith the removal of Python 2.x we can remove the unittest2 compat\nwrappers and switch to assertCountEqual instead of assertItemsEqual\n\nWe have been able to use them since then, because\ntesttools required unittest2, which still included it. With testtools\nremoving Python 2.7 support [3][4], we will lose support for\nassertItemsEqual, so we should switch to use assertCountEqual.\n\n[1] - https://bugs.python.org/issue17866\n[2] - https://hg.python.org/cpython/rev/d9921cb6e3cd\n[3] - testing-cabal/testtools#286\n[4] - testing-cabal/testtools#277\n\nChange-Id: I4a10dda03078c51380e8faad7684d2ba810f097f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7a5e8945f1ef70a76ba7de172c60b049e84c68f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7a5e8945f1ef70a76ba7de172c60b049e84c68f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
