)]}'
{"id":"openstack%2Frally~741162","triplet_id":"openstack%2Frally~master~I3a7cc7e279dc8aa2d78d93697de68f2cf8a62404","project":"openstack/rally","branch":"master","hashtags":[],"change_id":"I3a7cc7e279dc8aa2d78d93697de68f2cf8a62404","subject":"print completed workloads progress while running","status":"MERGED","created":"2020-07-15 12:05:18.000000000","updated":"2020-07-27 10:29:08.000000000","submitted":"2020-07-27 10:29:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741162-1595845748204-be75a880","meta_rev_id":"67c9f8b8fffcef12a8bac7bc789c469f50908745","_number":741162,"virtual_id_number":741162,"owner":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-27 10:29:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":0,"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":2,"date":"2020-07-27 10:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":0,"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":1,"date":"2020-07-27 10:14:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-27 10:14:04.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2020-07-27 10:29:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7bd27d26f395263a8386f497151a226a62f2b877","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-07-15 12:05:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8258ae2c86436830232bbba181a79b1158f0fbc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 12:32:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cad6cb0a6194544bed350fd8d31a4f7 : SUCCESS in 5m 39s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/3f2b8e224d154d588c5a288b55e99496 : FAILURE in 6m 37s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/20692a14bc21465b9535e337412e0b88 : SUCCESS in 6m 03s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ce64ac0a13d47af9798dd349105e334 : FAILURE in 4m 12s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/1883ef2f05a943c9b88233013633e4ec : FAILURE in 4m 25s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/16ba543d1fb547f3859a499eafb05b18 : FAILURE in 7m 46s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/7259043bac6344728d468df04bc7dae3 : FAILURE in 5m 03s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/dd6990f68fca46b291110e00176caf5f : SUCCESS in 3m 59s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/0989f1a4df7443e48f4e07d1c7358214 : FAILURE in 6m 51s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2f76165e7e9f488e8a237ba39ee06c7f : FAILURE in 6m 33s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/3097cb0e3e764946a18a4da2cec108f9 : FAILURE in 4m 41s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8cf938eb3324b61977a2bf0a7485543 : FAILURE in 3m 44s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58d144cea1f44597baafdaf625d13cf5 : FAILURE in 4m 08s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/ef44d799829949baaa7c58f6592c7998 : FAILURE in 4m 20s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/44335700169e47ca8b68748ff57e6c27 : FAILURE in 4m 12s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/6cf706afd1e840d295a027686ea1c8a8 : FAILURE in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2812be0dbddc298da9ac91177d4491077ff2a04c","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-07-16 01:47:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3e7d7e84baaf84cbbdd5cc22b3714d6a7f4f6d82","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-07-16 01:48:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"537a7dd39ce1e11d151a6d1bc65b2c8188bbc91b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 01:57:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f63bae2fe854142b0e51c36331fe74c : SUCCESS in 5m 18s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/1c2b498fc3074636ac1c019841d8cb06 : SUCCESS in 5m 34s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/9def323ac1014a63a5c57bef4ce7e1c4 : SUCCESS in 5m 18s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c5037bf63f040f68ed8ce94aaed0f1d : FAILURE in 5m 00s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/9a41128e0c5742ce9430f1421de4fe05 : FAILURE in 5m 29s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/8cb004b736624518835271360925745a : FAILURE in 4m 57s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/ee0797f1946b4f61bab2aa020aaa8545 : FAILURE in 6m 18s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/e9a3aceb201a4a3483306f787199ed7f : SUCCESS in 4m 01s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/55b7cb6e121d4867bceb8671e34b95c0 : FAILURE in 6m 47s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8da1a2dbeec5477e8cce38f973fe65b9 : FAILURE in 7m 20s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/530fe2f9f0f74ebda3ae5aeca103361d : FAILURE in 3m 42s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bab4d4aaa91a4db28cee67003ba67094 : FAILURE in 3m 52s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09144adf12eb4a9a94d8599542c9f31a : FAILURE in 3m 20s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/b1946b57858c4ee298cf1a1dc24b59ab : FAILURE in 3m 15s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/ffef42cb77df4c9cb98cbf3ce660f167 : FAILURE in 5m 26s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/7f989ccf42614268ad33e16ede368304 : FAILURE in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"600ecd75664bd51d113f9afe598e9c9c7fea22a3","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-07-16 03:52:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4a196b9b84647e0befd987e46e0e3fa6a80bab82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 04:06:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ae1b954b2334d30ac8bdc6103ea71f2 : SUCCESS in 4m 59s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/d2a608cba21a4fa58926fd63cee9c659 : SUCCESS in 5m 06s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/7605efdb506f4d3c9ca021dc7c966a21 : SUCCESS in 4m 26s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a0773dadb1644e2a1b9422522d28b4f : FAILURE in 5m 00s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/c852eaba203c4b41953d7a5e65cfd319 : SUCCESS in 3m 33s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/d29b7ab0c84e435a82a8f01939f26481 : SUCCESS in 5m 23s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/22ea5f70a6f84779af20e7bd70277fbf : SUCCESS in 5m 13s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/c5d56b6ec044429b8e08ce33032e57b5 : SUCCESS in 3m 36s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ed5165d006e44d6fb838ff8788cfa708 : SUCCESS in 6m 11s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/773108ed55a04f9b8f3a56222019100d : SUCCESS in 7m 30s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/d4e31109207c47798ebc0746d12f4c1f : SUCCESS in 8m 09s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7da0947eaa2e4fd7a936ba9b83fde6a0 : SUCCESS in 6m 22s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76810c4ad8424c24aa2b31fd49fa630e : SUCCESS in 6m 39s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/31e62dfa883843308aac8a636ac6eb52 : SUCCESS in 6m 57s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/3375ab2fb9b74581bff8c579b38dfe07 : SUCCESS in 7m 39s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/183f261d7b6c4af09f0cd47995fb58b7 : SUCCESS in 8m 26s","accounts_in_message":[],"_revision_number":4},{"id":"c07e6ef368c460aa64ea76dc9aa9f60bcec71465","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-07-16 06:43:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"98e8e1fa3706043325bc0c24627401886752a5ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 07:08:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2bb64a0602d4902a9041bf408af7827 : SUCCESS in 8m 42s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/77b6baf2d6934125ab43c060bd00ea26 : SUCCESS in 7m 47s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/81ccbd79b3df42a6905cb0d64c969e76 : SUCCESS in 8m 23s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a843437bf034d18a0948d609164cd34 : FAILURE in 7m 34s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/52c1ca8f1a184c9fbb858ab7716282ea : SUCCESS in 7m 36s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/7aa8aba6b5df4cc38cd6b04dc139671a : SUCCESS in 7m 17s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/b3befd64fb1f4086a1d385d43b56fdbf : SUCCESS in 8m 40s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/82238c14c3e64efcb6f6a2a67e1f641c : SUCCESS in 8m 06s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/46b74211eedb4fccbad3a3d76dede23c : SUCCESS in 12m 04s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/92adbafdbf6346fcb3d168c20637bd27 : SUCCESS in 12m 21s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/1900a4eb6a4144fd8617217c08f72d2c : SUCCESS in 11m 42s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f380534b4b5b4b00b16099d615f099d4 : SUCCESS in 9m 32s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4e29bd952ed4aa188ad3072fe935a4b : SUCCESS in 10m 40s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/52cdc5ac92af405c88897fad9a55f92d : SUCCESS in 8m 53s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/82b05ae095224ee6bc744e145d00eb53 : SUCCESS in 9m 57s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/390565e0ca3f4ffa9909e93cdaa2a693 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":5},{"id":"9a16fd9cfb156b169df17efcb6ebf4137b21a7ed","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-07-16 08:17:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4b7c11d048bc5db02ce38a6e88d438636b4a3f0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 08:37:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f78a71837f6478aacccdd6f5af0aba0 : SUCCESS in 5m 19s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/0bfc4a0ffe824c74b5e34914285fc6a5 : SUCCESS in 4m 49s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/c46434ba05a34ce4b108d09d43793176 : SUCCESS in 5m 00s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8d7e2feb7184f829b0decaad6ebae15 : SUCCESS in 4m 13s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/a1f519d316e64825b9c4cf3dc7c29d91 : SUCCESS in 4m 40s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/dcbbd39bdaf6400699167c77cf9504a4 : SUCCESS in 4m 07s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/a64fdaca736c4ae498e0d8af40c9e9ff : SUCCESS in 5m 07s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/38a57fc36e6b456c87de7725e9d7581e : SUCCESS in 4m 03s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/90cac07144a043f68f63816edfa2e57e : SUCCESS in 8m 55s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5fbbbf61089b4ed6877b51a5d87b4eed : SUCCESS in 7m 32s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ae1a043268ab4d479ec268afb3cc1889 : SUCCESS in 7m 56s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb305a77f0084ff2b6527b4858b9ac1b : SUCCESS in 7m 21s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c7aa896c87d4caab2251ef596e02543 : SUCCESS in 6m 11s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/83b72287e8d342f8937d7c01b0f08ae6 : SUCCESS in 7m 10s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/f3515df124e649ceb71e205933bd9f0c : SUCCESS in 7m 33s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/bdaf3825bc984e29963479d640e9924f : SUCCESS in 8m 33s","accounts_in_message":[],"_revision_number":6},{"id":"d89a36746143c21ab14b2f6d9cc4c448bf8ba85c","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-07-24 10:00:30.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b4a79e0578d456410d908b848423d9bfd462b7c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 10:16:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b31531beef894a76bcac0aaad11dc05b : SUCCESS in 5m 46s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/3455ddf5a30044ccb139959553bf5eb2 : FAILURE in 7m 27s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/1adf7dfe0dcb4b0ab531cab62da21dcb : SUCCESS in 6m 01s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/f928fedc768848ecb0d7d8e1f5219843 : FAILURE in 4m 20s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/e1e1fc69ec304d1a9c75d09c8ff52230 : FAILURE in 5m 03s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/cfb8e0b82a6f4afe9d36ef265dcc44c1 : FAILURE in 5m 52s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/ac5c43d849c74342a865fd674190c78d : FAILURE in 5m 05s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/fbf5858f434d42bfa75f388cfeddaa5f : SUCCESS in 5m 10s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/62cdf039f192491ba3340a73c7fd1ea0 : FAILURE in 6m 45s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/37e0a904edd3407d93dd97c1c2fdd811 : FAILURE in 7m 46s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/50a8732ad5fd4b73a69aec7168881b6f : FAILURE in 5m 29s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b6c1d0794024965b54809129366c8c6 : FAILURE in 3m 54s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebde1660d2a7405fa2fd6c44d680eaa3 : FAILURE in 3m 36s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/820f8a9ae0264d9e98f576b27dcf8109 : FAILURE in 4m 02s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/2575dae8e2e0484c85e97738ec78152b : FAILURE in 5m 23s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/8233950d6391471586f1594c1132d9b2 : FAILURE in 5m 28s","accounts_in_message":[],"_revision_number":7},{"id":"9154a747e21abccf0165f350c7ca5e1667db31ce","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-07-24 10:48:34.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9039d99c7eeb29ec6155bc8904dabcac45b0e13c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 10:58:12.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74207705dcf4be3be942023a3879bee : SUCCESS in 6m 05s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/43b2ac47da41454b80c305b51436d44d : FAILURE in 6m 05s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f4762f24e25f4122a9497e80a23e3587 : SUCCESS in 5m 51s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/8243f0c8db5f4cbd9553fb227d38aeac : SUCCESS in 5m 18s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/b7dc95ea513042a2ba82346ec3913872 : FAILURE in 5m 16s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/16558afe6e144fd6b78a926ad0e96f06 : FAILURE in 5m 00s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/2b95ae6d95094ca6bddcec9eca7beb2a : FAILURE in 5m 14s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/a1bd8737816b4e959d66e006a7f723d4 : SUCCESS in 6m 02s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/091a307d759a4513b804a3f9bff0b646 : FAILURE in 8m 04s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1d71300703114268bd47dd00cc0c55a7 : FAILURE in 8m 10s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/7dbb4df425824b57a7b785cbedabf750 : FAILURE in 5m 24s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12dad419c4de470e9b76840af8635e26 : FAILURE in 4m 27s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/021e53ec3bb14d3e8bd577c58920123d : FAILURE in 3m 57s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/bdadac09c99649e8922caca623c8825a : FAILURE in 4m 43s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/8a69558e70ea4f3cbc11035fb79267b7 : FAILURE in 4m 55s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/3e2a05584d69443c83507e0bc57d2199 : FAILURE in 6m 56s","accounts_in_message":[],"_revision_number":8},{"id":"143671d3b03c425db3af892bb12c37cb65f776a9","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-07-24 14:18:45.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f0b0f2e6bc87d21a2e67cbd8e5c28cee8a9892fa","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-07-25 02:20:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e12358ceb17b7c9e1a81dab28be1730bdb869e89","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-07-25 02:25:58.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d05410f01230f3bb7a0c863acb1c73deb3a0a914","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-07-25 02:32:39.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"777cd53874842177fcb8f255bcbfa67651e78f5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 02:45:42.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/706040cad7c3423fa54fb0e26d9a680e : SUCCESS in 6m 47s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/9f15acf316284e45a7fba3c7a145b703 : SUCCESS in 6m 36s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/dd09f9b133894c51b5555b2e976c9c25 : SUCCESS in 5m 38s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b2097d0ec204b08ad28790815e2d910 : SUCCESS in 3m 36s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/17e6ace64e9643a18be71dae9ad2f861 : SUCCESS in 4m 09s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/1089107fdda146e385c5ac82d7ec754b : SUCCESS in 4m 16s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/7fcc423e32c349028b23e3192c1d3b06 : SUCCESS in 4m 03s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/f88b2df36c864fe597363ecefba91867 : SUCCESS in 3m 33s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ef5631845f274a33abfbe7fb88617d46 : SUCCESS in 6m 31s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5e94cf4c46224dcfb2e05e434034c996 : SUCCESS in 6m 20s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/86380017c0834318b1a0416631f1423f : SUCCESS in 6m 58s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4dfb305341445b399db497f179f7d6a : SUCCESS in 6m 21s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93c1be6c8358453abe655235a51302ee : SUCCESS in 6m 04s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/0179fcdf616d49bba81467787c2ba594 : SUCCESS in 6m 47s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/7572cf54128f4752b8c6dd9ec836010c : SUCCESS in 6m 51s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/8a050d085cee41e8803fd8f053f77664 : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":10},{"id":"a21cfcb50c52a56b6f580e7d773de59a0b223f59","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-07-27 10:14:04.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"f52c3e874073835b3e91fdcd54235cae10f12402","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 10:14:17.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"ba9bf432c4674ed24bf2f9f8c77feee943a5e608","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 10:29:07.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebdc8b92d8b94a889b2517ed13230a16 : SUCCESS in 5m 51s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/28f5ea80a8b04e16919728c42fa42259 : SUCCESS in 6m 47s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/cda251259c9940289ccd6b303b1c536b : SUCCESS in 5m 59s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/16688c68053f4fdbb1789543ad64fbce : SUCCESS in 5m 14s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/80e211893ec14452b54467925b71e230 : SUCCESS in 5m 29s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/b8a4b123ec84455c99d1fc6f28fa5e76 : SUCCESS in 5m 01s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/fbf932fac3c548bc899c6bf873c97ba1 : SUCCESS in 5m 42s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/1c1167cbdfea401abac3544fd2c87ee8 : SUCCESS in 7m 44s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/581c351b99f04315b895417a0cae2ece : SUCCESS in 8m 14s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3dfae0a26ef4252b4af28f5e8f30ac8 : SUCCESS in 7m 31s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b990ed10e770430b89261838ddff6344 : SUCCESS in 6m 47s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/9a59603a954c4cc896bfd9d7f9a92dd3 : SUCCESS in 8m 17s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/478ae4c30dcf467884f36627c91bf062 : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":10},{"id":"67c9f8b8fffcef12a8bac7bc789c469f50908745","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 10:29:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"112f9ca3bbf8bf429e5a26db6c8106a07283cc84","revisions":{"57786576935fae5ef089022a5c68419a9fcb6a97":{"kind":"REWORK","_number":1,"created":"2020-07-15 12:05:18.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/62/741162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/62/741162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/62/741162/1"}}},"commit":{"parents":[{"commit":"695f5ecbabf1332da7efb38c0f21708a79bd5e52","subject":"Merge \"Switch from unittest2 compat methods to Python 3.x methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/695f5ecbabf1332da7efb38c0f21708a79bd5e52"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-15 12:04:10.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-15 12:04:10.000000000","tz":480},"subject":"print rate of workloads process while running","message":"print rate of workloads process while running\n\nChange-Id: I3a7cc7e279dc8aa2d78d93697de68f2cf8a62404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/57786576935fae5ef089022a5c68419a9fcb6a97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/57786576935fae5ef089022a5c68419a9fcb6a97"}]},"branch":"refs/heads/master"},"9c6882a3e63a8d19a50e75c75ed6bbfa7ccc2eef":{"kind":"REWORK","_number":2,"created":"2020-07-16 01:47:03.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/62/741162/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/62/741162/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/62/741162/2"}}},"commit":{"parents":[{"commit":"695f5ecbabf1332da7efb38c0f21708a79bd5e52","subject":"Merge \"Switch from unittest2 compat methods to Python 3.x methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/695f5ecbabf1332da7efb38c0f21708a79bd5e52"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-15 12:04:10.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-16 01:46:07.000000000","tz":480},"subject":"print rate of workloads process while running","message":"print rate of workloads process while running\n\nChange-Id: I3a7cc7e279dc8aa2d78d93697de68f2cf8a62404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9c6882a3e63a8d19a50e75c75ed6bbfa7ccc2eef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9c6882a3e63a8d19a50e75c75ed6bbfa7ccc2eef"}]},"branch":"refs/heads/master"},"9a4625c667120f9be16bdc20480ae7b05b40aa05":{"kind":"REWORK","_number":3,"created":"2020-07-16 01:48:37.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/62/741162/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/62/741162/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/62/741162/3"}}},"commit":{"parents":[{"commit":"695f5ecbabf1332da7efb38c0f21708a79bd5e52","subject":"Merge \"Switch from unittest2 compat methods to Python 3.x methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/695f5ecbabf1332da7efb38c0f21708a79bd5e52"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-15 12:04:10.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-16 01:48:03.000000000","tz":480},"subject":"print rate of workloads process while running","message":"print rate of workloads process while running\n\nChange-Id: I3a7cc7e279dc8aa2d78d93697de68f2cf8a62404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9a4625c667120f9be16bdc20480ae7b05b40aa05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9a4625c667120f9be16bdc20480ae7b05b40aa05"}]},"branch":"refs/heads/master"},"d226bcdf2a36e5de0441fa2867800d653638aa09":{"kind":"REWORK","_number":4,"created":"2020-07-16 03:52:53.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/62/741162/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/62/741162/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/62/741162/4"}}},"commit":{"parents":[{"commit":"695f5ecbabf1332da7efb38c0f21708a79bd5e52","subject":"Merge \"Switch from unittest2 compat methods to Python 3.x methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/695f5ecbabf1332da7efb38c0f21708a79bd5e52"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-15 12:04:10.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-16 03:52:07.000000000","tz":480},"subject":"print rate of workloads process while running","message":"print rate of workloads process while running\n\nChange-Id: I3a7cc7e279dc8aa2d78d93697de68f2cf8a62404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d226bcdf2a36e5de0441fa2867800d653638aa09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d226bcdf2a36e5de0441fa2867800d653638aa09"}]},"branch":"refs/heads/master"},"3969957116a1f6f3da443c7310ee0ec9066fcc23":{"kind":"REWORK","_number":5,"created":"2020-07-16 06:43:28.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/62/741162/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/62/741162/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/62/741162/5"}}},"commit":{"parents":[{"commit":"695f5ecbabf1332da7efb38c0f21708a79bd5e52","subject":"Merge \"Switch from unittest2 compat methods to Python 3.x methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/695f5ecbabf1332da7efb38c0f21708a79bd5e52"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-15 12:04:10.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-16 06:42:17.000000000","tz":480},"subject":"print rate of workloads process while running","message":"print rate of workloads process while running\n\nChange-Id: I3a7cc7e279dc8aa2d78d93697de68f2cf8a62404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3969957116a1f6f3da443c7310ee0ec9066fcc23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3969957116a1f6f3da443c7310ee0ec9066fcc23"}]},"branch":"refs/heads/master"},"b4e433f474a441fff3b0e5896249dff88343eb10":{"kind":"REWORK","_number":6,"created":"2020-07-16 08:17:15.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/62/741162/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/62/741162/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/62/741162/6"}}},"commit":{"parents":[{"commit":"695f5ecbabf1332da7efb38c0f21708a79bd5e52","subject":"Merge \"Switch from unittest2 compat methods to Python 3.x methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/695f5ecbabf1332da7efb38c0f21708a79bd5e52"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-15 12:04:10.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-16 08:16:21.000000000","tz":480},"subject":"print rate of workloads process while running","message":"print rate of workloads process while running\n\nChange-Id: I3a7cc7e279dc8aa2d78d93697de68f2cf8a62404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b4e433f474a441fff3b0e5896249dff88343eb10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b4e433f474a441fff3b0e5896249dff88343eb10"}]},"branch":"refs/heads/master"},"666348ca490823b5942c458071afb73ab6f2d759":{"kind":"REWORK","_number":7,"created":"2020-07-24 10:00:30.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/62/741162/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/62/741162/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/62/741162/7"}}},"commit":{"parents":[{"commit":"695f5ecbabf1332da7efb38c0f21708a79bd5e52","subject":"Merge \"Switch from unittest2 compat methods to Python 3.x methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/695f5ecbabf1332da7efb38c0f21708a79bd5e52"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-15 12:04:10.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-24 09:57:42.000000000","tz":480},"subject":"print completed workloads progress while running","message":"print completed workloads progress while running\n\nChange-Id: I3a7cc7e279dc8aa2d78d93697de68f2cf8a62404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/666348ca490823b5942c458071afb73ab6f2d759"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/666348ca490823b5942c458071afb73ab6f2d759"}]},"branch":"refs/heads/master"},"c9d7da89df05742632c9866f0991222c102db88e":{"kind":"REWORK","_number":8,"created":"2020-07-24 10:48:34.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/62/741162/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/62/741162/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/62/741162/8"}}},"commit":{"parents":[{"commit":"695f5ecbabf1332da7efb38c0f21708a79bd5e52","subject":"Merge \"Switch from unittest2 compat methods to Python 3.x methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/695f5ecbabf1332da7efb38c0f21708a79bd5e52"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-15 12:04:10.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-24 10:47:57.000000000","tz":480},"subject":"print completed workloads progress while running","message":"print completed workloads progress while running\n\nChange-Id: I3a7cc7e279dc8aa2d78d93697de68f2cf8a62404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c9d7da89df05742632c9866f0991222c102db88e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c9d7da89df05742632c9866f0991222c102db88e"}]},"branch":"refs/heads/master"},"a499be0f328e5ed78088e9f39e089b78c88a7517":{"kind":"REWORK","_number":9,"created":"2020-07-25 02:25:58.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/62/741162/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/62/741162/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/62/741162/9"}}},"commit":{"parents":[{"commit":"695f5ecbabf1332da7efb38c0f21708a79bd5e52","subject":"Merge \"Switch from unittest2 compat methods to Python 3.x methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/695f5ecbabf1332da7efb38c0f21708a79bd5e52"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-15 12:04:10.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-25 02:23:24.000000000","tz":480},"subject":"print completed workloads progress while running","message":"print completed workloads progress while running\n\nExpect to see the completed and uncompeted workloads count while rally\nis running.\nChange-Id: I3a7cc7e279dc8aa2d78d93697de68f2cf8a62404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a499be0f328e5ed78088e9f39e089b78c88a7517"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a499be0f328e5ed78088e9f39e089b78c88a7517"}]},"branch":"refs/heads/master"},"112f9ca3bbf8bf429e5a26db6c8106a07283cc84":{"kind":"REWORK","_number":10,"created":"2020-07-25 02:32:39.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/62/741162/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/62/741162/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/741162/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/62/741162/10"}}},"commit":{"parents":[{"commit":"695f5ecbabf1332da7efb38c0f21708a79bd5e52","subject":"Merge \"Switch from unittest2 compat methods to Python 3.x methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/695f5ecbabf1332da7efb38c0f21708a79bd5e52"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-15 12:04:10.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-07-25 02:31:55.000000000","tz":480},"subject":"print completed workloads progress while running","message":"print completed workloads progress while running\n\nExpect to see the completed and uncompeted workloads count while rally\nis running.\nChange-Id: I3a7cc7e279dc8aa2d78d93697de68f2cf8a62404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/112f9ca3bbf8bf429e5a26db6c8106a07283cc84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/112f9ca3bbf8bf429e5a26db6c8106a07283cc84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
