)]}'
{"id":"openstack%2Frally~745790","triplet_id":"openstack%2Frally~master~Ifebef92bc0f549e4d0c778261bfc9b40e90d39af","project":"openstack/rally","branch":"master","hashtags":[],"change_id":"Ifebef92bc0f549e4d0c778261bfc9b40e90d39af","subject":"restart the task","status":"MERGED","created":"2020-08-12 06:20:32.000000000","updated":"2020-08-28 11:57:30.000000000","submitted":"2020-08-28 11:57:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745790-1598615850420-1d2acb16","meta_rev_id":"1e304b5b85954f7518d0ff4f0b35b5eeb0d3c319","_number":745790,"virtual_id_number":745790,"owner":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 11:57:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-08-28 11:42:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-08-28 11:42:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-28 11:42:18.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2020-08-28 11:57:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c877b634d7b716ef142caf18d796a6ba21f7ed51","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-12 06:20:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ab975fbc6d9b0fde463a318e60e6c0f131480ee","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-12 06:33:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"57eeb7b118fd8e79b968d5226fd5f79d3512fbc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 07:17:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6c0b86edb04b51b28582cb83b03b5b : FAILURE in 5m 26s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/80c26e6e99ea46a78a6746351e0068e5 : FAILURE in 7m 22s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/33e2fb3d9868429da45d010a746da21f : FAILURE in 5m 18s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/d809af50305b4c2893949e587807c1a7 : FAILURE in 5m 16s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/a94b275b0c044ae2bd90e342f36cb446 : FAILURE in 5m 51s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/4bfeca0cfc9749ebabbdcddb6be7052d : FAILURE in 6m 45s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/cb9c61dd76e846b7874d9d8cf3c0afd7 : FAILURE in 6m 18s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/bba2f5aaf8a34d5598830aad75307f25 : SUCCESS in 4m 11s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/3ad4642632e2480f94d598dda5cd41f0 : FAILURE in 5m 04s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4c79835a3bb24daca9d2d4ee936b418c : FAILURE in 5m 28s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/4e2e7d7607c345d0bdcf64ecfa658479 : FAILURE in 4m 40s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d9b00fdc05a4ca8b0e63060e305eb60 : FAILURE in 4m 04s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b94fd2ba56e3409b84f41fb72c9ac353 : FAILURE in 3m 53s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/22df4f48420e4d6980e414d5d6194693 : FAILURE in 4m 00s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/a86ce12628314cabba3cce2f7c64e10e : FAILURE in 4m 27s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/7e05f9e8779a4c02a85de965bc6cc53f : FAILURE in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"2f1d32afaa740d6608c41894628fbd0ae8d03819","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-13 03:03:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aff741f831336bc12dd5aadf4a09a088b1ac2c82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 04:05:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c1cac529304e79a9837d6a328e0622 : SUCCESS in 8m 09s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/cd31649ce04e4deea65b5a809b1c065e : SUCCESS in 4m 11s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/2ef97edbfe1748c5873165798b1d7a41 : SUCCESS in 5m 07s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bbf0491a6244c0bb4e4cfe4c56801af : FAILURE in 3m 34s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/f6240b9482cd421883768b5d0a4427b5 : FAILURE in 7m 24s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/0e38613f8b244d5fb873390f8aac1be9 : FAILURE in 3m 39s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/a1e6188a91314bdcbb4a029c8983b285 : FAILURE in 4m 48s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/913337f56a544483ad0aacd3eb34cd96 : TIMED_OUT in 31m 35s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/48aaaea4476c4b5fa114f7a98a1d3e32 : FAILURE in 8m 03s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d6fe5fbf09044081b171296e8bd94e36 : FAILURE in 6m 58s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/b3de50642ce24fa1a6ba0ac18eaba787 : SUCCESS in 7m 14s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48fc773753664b34b1baf2da3b72f41b : SUCCESS in 6m 35s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9223325ee7c943ca95a0799174e94b3e : SUCCESS in 6m 28s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/9123363b51e44a2b84d77051516b9931 : SUCCESS in 6m 54s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/2d1f3f610dff45598c661c463664fff8 : SUCCESS in 6m 50s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/955f51f947be4e7bb95452e337fc6bbc : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":3},{"id":"d83ee7c596152522f9dcac6c9553dee5875ec666","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-13 06:28:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"19a0a8963c926df72aa782385539d719a2ebd7e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 07:24:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ac38e503fa4ad1aa1ce72ae93ae139 : SUCCESS in 5m 34s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/2c1ba8960c3f4f26a7266161f2f2435c : SUCCESS in 4m 50s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ffdf2a69fac6407984802b71a960c727 : SUCCESS in 5m 37s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/181db081a0804898944b62faa816bf7f : SUCCESS in 5m 08s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/6e927c0f91cb45268345cfa907d5165c : FAILURE in 4m 47s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/fd6b4e08f5e949dca9b0da3ad3ed28e2 : FAILURE in 5m 03s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/e8b2fbf38fd74bfdb65a3884097460b2 : FAILURE in 4m 15s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/95669161ff884e0098124beddd0b86d5 : SUCCESS in 4m 03s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/8e37531c8c254f74800061d5e77360c9 : FAILURE in 7m 20s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7f054c265a1d41d5aa25b2df23db47df : FAILURE in 7m 22s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/6c64947bac6f4bf38af3af4a854b0917 : SUCCESS in 7m 11s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1502a70b7ff5437fb7725f4761c3fa56 : SUCCESS in 7m 48s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/603d6ff954c84403981b5d8441a119e5 : SUCCESS in 6m 51s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/5c16c2f64c8f4c76a90504d21a791924 : SUCCESS in 7m 18s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/ded29c270efe4bd6917e2f907a98e364 : SUCCESS in 7m 03s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/8fabebbdfe5149ecb174b2c91896c0a1 : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":4},{"id":"edd5dfe9f571dd0448aa49fade2bcd1d9f5fe2d7","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-13 08:59:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2d8cba02f1372a012f4cb09d54c0b7e630747e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 09:29:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f95bb43aade44804aff780656f0a1f5f : SUCCESS in 5m 43s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/d09ce44d7b6a457584b70d8c5c570cfa : FAILURE in 6m 42s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/186fb4df29c44807b82b0bdad64a3f47 : SUCCESS in 5m 02s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a183fa0ac54409cb9de1834c793ee65 : SUCCESS in 3m 34s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/5b9068f398a44f7192fae34594da1e40 : FAILURE in 4m 50s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/9e3b836bdcd14984bbbc30685838b53f : FAILURE in 4m 45s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/578affa5cd8741588547c052872a4ba3 : FAILURE in 4m 42s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/91c1f1c354da4467b8534c8748430b77 : SUCCESS in 3m 54s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/a03677b5da0d44b68447218f6e05f32b : FAILURE in 6m 53s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bfa4d32f5212419fbda7704b37e55e9f : FAILURE in 7m 32s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/492d8cbf9dcc4f04928326c6dd1e4f30 : SUCCESS in 8m 04s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf7f241f8dfb435cab546493dba3a92d : SUCCESS in 6m 48s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21293e95d1304e5f8e9f593610b062af : SUCCESS in 6m 58s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/7712c3d2b94a4c178fdc950104ce8b58 : SUCCESS in 7m 03s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/bde98ab9aa144030bb5f33ca931fe2b5 : SUCCESS in 7m 16s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/a831e7f9dbce41278eeb78f8643df25f : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":5},{"id":"82a8e8974a89806a928921a6b5fa293d221f0918","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-13 10:55:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f19607335b0c6d9d8524d04ed386996779d678c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 11:37:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da2fe5055c1643e9983650024f1ece42 : SUCCESS in 5m 00s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/9f5b92640200451ea3b3e18f2975e041 : FAILURE in 5m 13s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/fc1212f651c34f469f1c03f50fe37471 : SUCCESS in 4m 33s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/91de19737930457a87b2f7279c361439 : FAILURE in 3m 47s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/f1ca67a21c2c4e3bbd1f05ccd3a0d48d : FAILURE in 4m 38s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/031aebbc8c974a47be2d1fde64a36b95 : FAILURE in 4m 17s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/f41cc07cb42f48cca52ba5d5c45b5ce0 : FAILURE in 5m 43s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/37ac6689144f41f0aae9490bb8287a26 : SUCCESS in 3m 43s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/8c572a94f9504b739ee7963952290893 : FAILURE in 7m 56s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e5b11908fb3c448da7fa2cd5b12e93ba : FAILURE in 5m 27s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/b60ca895b28f4d7797547f164b2cda27 : SUCCESS in 6m 46s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/beb0f4d952e94a08a8fd4f0ab583d0d0 : SUCCESS in 6m 06s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a59006f490aa46589b11f3e4664c5b80 : SUCCESS in 5m 55s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/34175889b06348019ef639e909e39832 : SUCCESS in 6m 38s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/2adbd40ba51949969aa5a604aea07c95 : SUCCESS in 7m 05s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/509d241643db492b8fbd6b3fd303617f : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":6},{"id":"cc819bdaff9ae59a756bb6cd48f19e724a4be87e","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-13 12:04:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1932b2a6f317a4197449996590541d92249fd07f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:19:46.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9e9dc789904a14a301cd04e10094e3 : SUCCESS in 5m 21s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/2357f06d71bc4e3b8b0c7feeddd3d6ca : FAILURE in 4m 54s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/971d44219aa346f388886755efdeba5d : SUCCESS in 4m 39s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/8190399f59e14153abc279828f3a6d8e : FAILURE in 3m 59s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/cdf16bdead224992a6580df2db387f9c : FAILURE in 4m 08s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/1fafb647533d4ee29a212d8149981ff1 : FAILURE in 4m 03s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/8e28f3db31b84a81b7eebd8f0db0a491 : FAILURE in 4m 13s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/15471aa7477145a3b236df23e4a80894 : SUCCESS in 3m 31s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/dc132b7c4f324462ba331106b4f1bcce : FAILURE in 5m 47s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/21365616ab8145e8960387b3afdcbd4d : FAILURE in 5m 32s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/959d85451f0648968214913bb2835332 : SUCCESS in 7m 10s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d339cc31107e4722a016def26768086b : SUCCESS in 6m 46s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc4cac25273248759b5a5ba48ae2f4a3 : SUCCESS in 6m 44s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/0ae02a4d0dd34ce3b0a4f48c42e7a15c : SUCCESS in 6m 45s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/927dc19fcc074908b52d28891d0e6857 : SUCCESS in 6m 40s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/789992fb215c4f2da7d36005cb333aa8 : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":7},{"id":"9551a810e1e7cefe68602b0de0cfe77b98d7ba9a","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-14 01:17:01.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"93b3cd14cd75fa8c67bc328feed6e087dcc69e6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 01:27:45.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cbe360d07344145b5dd5d28fce199b6 : SUCCESS in 6m 54s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/adf9fa8e034b45b4be317b4a55bcaf60 : FAILURE in 4m 06s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/537b44f12b904acb9b23fcc7572054ee : SUCCESS in 4m 35s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/80969609a4ef4ae8bdd7caf4db6d5cd6 : SUCCESS in 5m 02s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/f50002f6f24345cd92faa8f7c28d16c5 : FAILURE in 3m 31s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/ac2d711eaa094b6489479b87b60a0267 : FAILURE in 4m 24s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/f9f5c6ad59a74365b6ed4e585ac2014f : FAILURE in 6m 10s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/64fa007e1ba444228190872703f5ed75 : SUCCESS in 3m 10s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/a396a05a599a4b0aba888c0db11dd68f : FAILURE in 6m 02s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9a58ec5b13844802954444b498cd4a4d : FAILURE in 6m 10s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/4f821f9f98af4221b59342c2721ced16 : SUCCESS in 6m 33s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79934ced59074f8cb45ab93e1e16b0ff : SUCCESS in 5m 54s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e68887798da43f3bf57796587ab82b7 : SUCCESS in 6m 44s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/a3c1489d7d9b4d8bb4f0990ff665e17f : SUCCESS in 6m 01s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/8f726651202848a1b43bf7fa04289b63 : SUCCESS in 7m 18s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/7b87f53c58ec4c4791b6b1f096cc2286 : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":8},{"id":"f09e5cd97f6ad93922ac94129a8a8a6b0ac83395","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-14 02:01:27.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b84114df775c31205d7c9b3fcf37c31c02204181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 02:13:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa70a405d51b4cd18f684d12ad8e5504 : SUCCESS in 4m 15s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/c8a0f5fa71d94e93bb0b68b48c9608cd : FAILURE in 4m 50s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/c318cff386e844c18a116200cf723447 : SUCCESS in 5m 00s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/89ce5e86c2554eebb3dd0c8296436b63 : SUCCESS in 3m 40s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/b2a4b730a65147eca741c78033b567d7 : FAILURE in 4m 29s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/184d0bc59b7c4b87b5caae017fd89564 : FAILURE in 4m 11s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/fe8fd0aec8354034becbd0172a81ac6f : FAILURE in 5m 09s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/4fc4a530070f4a95b7889526444a1834 : SUCCESS in 3m 17s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/835a7d8ae60d4c7fb3c3cf4b213e92cd : FAILURE in 6m 07s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/83719310053644d5929d1c31403caa79 : FAILURE in 6m 15s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/7346656b2c044a3ba6a83c0cf3664938 : SUCCESS in 7m 19s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a138e9d8f9904396ac94d44757ad0a0e : SUCCESS in 6m 39s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4081a79915174812a160921a011ed50e : SUCCESS in 6m 20s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/26e4e4fc9f1f40c884a2b5a71a162108 : SUCCESS in 6m 53s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/23270517084340ea922d717764cb8610 : SUCCESS in 6m 33s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/67469ccf32944484b2b6b7ce2f6a1035 : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":9},{"id":"038cb5b0bb6b0d92f6039e632c4c8de31e8bb15e","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-14 02:29:09.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c41f7e6653a693d1896efe8b11c3ce26f8489a35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 02:43:07.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47ead1da5aed4354bfc04e33e28ece90 : SUCCESS in 4m 46s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/04cfa6d1e7d7438d9c8828b89a6d303a : FAILURE in 4m 53s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/5372f48f168c452a960e257359f13019 : SUCCESS in 4m 31s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/c94362d1913045c9ba74d95f3eeb62a4 : SUCCESS in 4m 27s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/ca024fdf95a84c2dbfa5cb30ef297d1d : FAILURE in 4m 06s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/fa8c558b7c154845a77aaa223ed65a8b : FAILURE in 4m 38s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/c7ae6f8528304e7fa72f926dc622d3ca : FAILURE in 4m 19s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/ab9f84fd963f4d8894da76388db513b8 : SUCCESS in 3m 37s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/c6db20cb3ce34ed1b28d2717dc69793c : SUCCESS in 6m 15s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1176c75cb3c44e8babe6e24c2ce6a80f : SUCCESS in 6m 50s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/09b26f6044a84d2f80d3f20353522a75 : SUCCESS in 7m 02s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86cf57da62b34441b80dd2319b2db501 : SUCCESS in 5m 50s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b84a49e4135142129de92e7f219686c1 : SUCCESS in 5m 50s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/66735533574e4c8e9efb6183d24ea9fa : SUCCESS in 6m 59s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/d4d30b3d67674735a2f04913bc016569 : SUCCESS in 6m 45s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/b5678775942442bd8c3bcc19a5b8bea4 : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":10},{"id":"6fc3a2bc93d5458320d63df7aa1eda5de615462a","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-14 03:26:11.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"73f8cc6bd2b714568d2a1ac1acd8987035e69147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 03:37:05.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a422bae22474c23aaf355b53e63c638 : SUCCESS in 4m 59s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/8d12fc0abc19463db4d40a1960e116c3 : FAILURE in 5m 20s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/42a7d9f7bb8f46bfb5542a88aa9fb596 : SUCCESS in 4m 20s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1d0ec61b971485db6afdcdfc4857683 : FAILURE in 4m 01s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/a08ac555264948678b6da335e0175ea1 : SUCCESS in 4m 09s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/5d447af88c404b728660aed5cff8f0ae : SUCCESS in 4m 44s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/f167718eca964cd29843f2c0115d3b57 : SUCCESS in 4m 19s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/7b79355ed305495cb45d379c12cb0d3d : SUCCESS in 4m 16s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/9f579dc229814e1c8fe1f2b7c5a254f1 : FAILURE in 6m 07s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ff3f873824bb419d8674e9061fd86fd1 : FAILURE in 7m 04s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/fdc1300938a14f3ea7c1d3664c597024 : SUCCESS in 7m 10s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ec3656293cb42ff939ce78a5cf2113b : SUCCESS in 6m 06s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e70d53d47354ba6a8aa911ff33b8fad : SUCCESS in 5m 45s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/f4fb49d2f3aa4f5ea32d6627e155941b : SUCCESS in 6m 30s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/c720e456bf084d338d5acc487aa5240f : SUCCESS in 6m 30s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/193542bea17e46dfb70404ae794fbd28 : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":11},{"id":"a882d3a7dbc38bb07fe5d0b5ec7f7d9db5ee6743","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-14 05:50:28.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"58a5e4106678fbb30984e9272aad66d43855706d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 06:04:29.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f8cce806a264d73b121c9db1df6f790 : SUCCESS in 7m 29s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/f385e8d424d24a129254b6f029d6706a : FAILURE in 7m 21s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/7f57dc4690a64f378bf72f8f956cbff3 : SUCCESS in 6m 49s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/128ed2aa623c4b3bb7e5ef47ece5dd8c : FAILURE in 5m 54s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/e540fbff802f4af686297573dd9e0774 : SUCCESS in 6m 21s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/966b20a5cae642c69b36e9b8f38b2aea : SUCCESS in 6m 52s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/099fcea3b5d744239c43b69355bf54bc : SUCCESS in 7m 01s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/38634417ebe540118623b7765f3278d8 : SUCCESS in 5m 55s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/70ced030191e46eaa260ca9f1277c141 : SUCCESS in 5m 47s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b264542448554c85a642f90fa14813e0 : SUCCESS in 12m 16s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/360383a60ee34d529e9711633f5725e1 : SUCCESS in 9m 46s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d258df32bad4d17b4f914e1a9a52910 : SUCCESS in 8m 20s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/609293b51727469693d355f7abea05dc : SUCCESS in 8m 56s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/8e9bde93e8e34475b156bb11ba1ccff0 : SUCCESS in 6m 35s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/69849714008f4dff8582542736a8cd3b : SUCCESS in 8m 41s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/8bf4f8273b65405b8d7065d58ec75eae : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":12},{"id":"6a7bc85eb4d6469ca38492c60731d03b2c38e184","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-15 06:46:23.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a638210d91e238de3d83afad71b5e9b14a7022fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 07:10:35.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/276c8567191a4aee831d398a6579ec30 : SUCCESS in 7m 17s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/3adb94aba1bf4558975b2b37d744ff5d : FAILURE in 6m 19s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/02fc23cbfa3d4ee397b452211aa893f0 : SUCCESS in 4m 42s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/67550a75294540f784bab524968e17cd : FAILURE in 3m 22s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/c5e43b0fa52a429ea86f1006a091c8b5 : FAILURE in 4m 50s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/9d7801e6a966476ca6755aab374f1f81 : FAILURE in 4m 23s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/9e5a853e8f9f42bd8bcc831523bf5104 : FAILURE in 4m 22s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/0d0270e5907d4b03b34138cc20def903 : SUCCESS in 4m 24s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/eeefb7d358dd4a3a89242efce10f55cc : SUCCESS in 18m 39s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8624fa500ff74926897e52e61464b985 : SUCCESS in 19m 01s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/dd2a75529cac4c44b8ccc24fdc522117 : SUCCESS in 6m 48s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/561729f9c72d4a20b97667c2eb81997e : SUCCESS in 6m 11s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e985af72a0346a6b2fec6d653d86921 : SUCCESS in 6m 15s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/49b41763f41c40adb418d4cd5d6be930 : SUCCESS in 6m 20s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/5ffe5704a1634506a74419a07cc22ffe : SUCCESS in 6m 49s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/51d3c54e0d9c431d882836c49979c10a : SUCCESS in 8m 33s","accounts_in_message":[],"_revision_number":13},{"id":"9e27fa3990e88ad0cfa7dd3a14e23d88595a25fc","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-15 08:22:12.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"d2cb1a6f753b0b3ed82c4d97db094a184a01b2a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 08:42:24.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8884c89009c40c48488754257d0cec7 : SUCCESS in 4m 50s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/bdc9072aad6644d3a0396f8184bff276 : FAILURE in 6m 17s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/c0621a3f423446babd6bed524c970c4c : SUCCESS in 4m 39s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/cac6f5591dd5416f921757652aee1086 : FAILURE in 3m 42s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/9027aaec3f70480bb2f13ccde881c2ae : FAILURE in 4m 03s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/80cac3bc310a4198884078573ee6bfe4 : FAILURE in 4m 25s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/eb6e72a46d5d4ec597d79d5f1c0f1c68 : FAILURE in 4m 38s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/6a3ef5f3510943f09f30fe863cfae743 : SUCCESS in 3m 38s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/55cdadeba2cd4eedb209077666300a01 : SUCCESS in 18m 30s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b4c319d3c39446e0a190cdc411da4cb0 : SUCCESS in 18m 46s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/00c2203c9db244e192b13826ae0d4d16 : SUCCESS in 6m 46s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa48a9ede841421b984aaa1153838763 : SUCCESS in 6m 11s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0118f55ff58f40f79413761af044db3c : SUCCESS in 6m 06s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/dce927b9a7a04f9488ae284f718aba52 : SUCCESS in 17m 29s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/f91bd217817e4780b608932487b8a124 : SUCCESS in 7m 04s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/ded1974729964588a3ead445b6b32a5d : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":14},{"id":"9eec7c512ac9ad68f8f6a5e760d21e3c65102481","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-15 09:10:59.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"ab150343e8b279dec9296e3bc554145314caa374","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 09:26:30.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9534a69280774acbaabddb2b3345acff : SUCCESS in 6m 41s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/d3fb38e7cc99495cb2f4ad56017ab0fa : FAILURE in 4m 40s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/825ae408a0504109aa6978840ee5fa98 : SUCCESS in 3m 21s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/41ec1a0d5c4144d1b1ccf678f0b2b401 : FAILURE in 2m 58s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/ed7a207a801e48babfaa98e7f7f92b12 : FAILURE in 3m 47s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/cf184c1a22d641c9bb343c5f5a616558 : FAILURE in 3m 40s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/1582fec16ab5440d8853c6d66a61f59e : FAILURE in 3m 51s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/8727cb38579041c6920bd46b6f050b08 : SUCCESS in 2m 56s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/c7ed571dcfb746dd92a750c659273525 : SUCCESS in 14m 04s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/24d73892439445dcbc5395a3327fa7aa : SUCCESS in 14m 21s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/f90d6acfea094f5a8ceab3c5e2e6af2f : SUCCESS in 6m 08s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f40f60e8c11411c924c0c184510c1b1 : SUCCESS in 5m 36s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/131017fff41c400cac9bc51887ddcc0b : SUCCESS in 5m 33s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/741decb0b03e46e4b8d22f65636709b9 : SUCCESS in 5m 49s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/41836d0f65734131bbffa6d580bf5d39 : SUCCESS in 6m 13s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/7102e256305a418797afe65272737e0d : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":15},{"id":"9196fe5ba688858771a04a71c78bcbaed229ba8b","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-15 09:35:14.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"6b461e4a795713f798b9333706618124330dab78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 10:00:06.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f57720786654e5aac463b349d3a5e46 : SUCCESS in 4m 09s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/2489801b17144954a2043ede0847a78a : FAILURE in 5m 41s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/4d8d7b2d3eee41bd88d0204a78a7c85a : SUCCESS in 3m 44s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dd81b7ba76d406aa94697ca26979660 : SUCCESS in 3m 25s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/bad6b43bccc64d02b2add7ffc28f8034 : FAILURE in 4m 02s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/2926b611953f4540b85ad794d99d0d96 : FAILURE in 4m 03s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/30d0e6c9509b469f8eb2c788ae00e142 : FAILURE in 4m 15s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/603e2748e0f94d389c69a66b668e57ca : SUCCESS in 3m 14s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/9dea4551eb8e455eb354a3dca26b7eee : SUCCESS in 18m 54s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/97b0d8e4829847748d2300bacc2564a0 : SUCCESS in 19m 04s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/c4fa8597df77415193cc7f01fce9d754 : SUCCESS in 6m 27s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b96b41ef958c4cb3a7ad86c504b6eac6 : SUCCESS in 5m 54s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cba7327c207045a7895b7e0a72310523 : SUCCESS in 5m 51s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/dab2c3c4523b48d0b730e35ff96e93e7 : SUCCESS in 6m 27s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/880e28e14ebf464f86e66e9ed1056a7f : SUCCESS in 6m 28s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/83ebc525597e4a66b34cd37c668daf1f : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":16},{"id":"1ccbaff6a8636f9450b6ab8954edba4d848c6a4f","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-17 01:28:10.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"2e896ac35658ee7df06c4a9b3900fe1479bef19d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 01:48:32.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba63781c141448509f7d200e75744dd9 : SUCCESS in 4m 40s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/8d97fde0734245979fac7e819045a4eb : FAILURE in 5m 36s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f1d3349618114fd9b52d411dd77a4896 : SUCCESS in 4m 05s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/081d29dbc6bd473f8236c4655c81237b : SUCCESS in 3m 37s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/cb5e1ffbfcab404aaa3e454eea24bdd8 : FAILURE in 4m 09s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/e5b653bdc9d546a38888db9552f29365 : FAILURE in 4m 38s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/cc0f359ccf774e53a96a83bd65f06017 : FAILURE in 4m 45s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/f50b22fda43e4d2dbc04cba4014168ce : SUCCESS in 3m 26s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/3da47db3d37240db947e12a8eb25f2c4 : SUCCESS in 19m 11s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/849fd7e43952455baf74fbb88dd778ca : SUCCESS in 13m 57s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/5ebca634d2c14c38b2f93307afe1a69f : SUCCESS in 6m 41s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf7011e6e811457584b9acff7b5ae431 : SUCCESS in 6m 07s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ed01ad514ae4b789e1d6129d6b21854 : SUCCESS in 6m 01s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/4358b77087f24d56a94304e0c5cd5073 : SUCCESS in 6m 14s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/9faba9b8108d4738a898b6fba2c0a729 : SUCCESS in 6m 51s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/ef26e25d8e784a539e8348453ea932b9 : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":17},{"id":"458d7a6bbd037dfe2d0aabab3732e9fc01efcba9","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-17 03:36:27.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"d06f74fc00fac409133802c43953d50c835a4efb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 04:01:31.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cad789fb02904b8a85148b7b9ebc5818 : SUCCESS in 4m 45s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/d183c5380cdb43fe92d865ae06f34693 : SUCCESS in 5m 47s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/4c86162159f64cce9807f695b4ce766b : SUCCESS in 4m 17s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeafdc8e0cd94ee8bcfd6c634e753f11 : SUCCESS in 3m 11s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/605bc5cfbdab4369b3caa13f01029356 : FAILURE in 3m 44s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/b7dc2b2b91e644c4a0003ac064def86a : FAILURE in 4m 54s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/39c8bc689ebd485996edab02bc1b1629 : FAILURE in 4m 37s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/6ab03fb107ca460a854d278bc590ad57 : SUCCESS in 3m 40s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/e96c9fd9e0c84c07a3af3e3ea8fa6be8 : SUCCESS in 16m 08s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3153a24740a24231a93333ad14e174af : SUCCESS in 19m 06s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/0ff22a779d0a4ddcacd70de69620ab5c : SUCCESS in 9m 34s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c5cf05d6acd4b018c4e962effa5fbfc : SUCCESS in 6m 01s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d995b16c48ed485a8c7f28fb2edd96ce : SUCCESS in 6m 22s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/cd2361db9a344b79a7cbc21fc47b1c0c : SUCCESS in 6m 42s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/ec22202d98334a86886540a65370d3ce : SUCCESS in 6m 25s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/aaa952814a224fb1862d534057b9b094 : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":18},{"id":"8d0f7733b173598f9c0333eb3f34ff5ac98cfaea","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-17 12:20:17.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"d5ad0f8a9462c9dcd76dbf6e98cf719783e3351d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 12:47:48.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a4dff2c9d324dd79bcabda644d7dd76 : SUCCESS in 4m 18s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/2742656bcad04b3baeea8bf094113226 : SUCCESS in 6m 07s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/e4ab94f11d614bf98a3eb3d8d185f270 : SUCCESS in 4m 01s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/93cbf38b7cb54987830cd3d0050ed686 : SUCCESS in 3m 12s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/b69fa034558e473781ef2607c34a0e39 : SUCCESS in 4m 04s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/c579599c854745d1a3adfedf3dc02570 : SUCCESS in 4m 21s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/cc73faed2c874bd9b7cfcb0d3ba9ce46 : SUCCESS in 4m 07s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/7a0c1d795800456abaa5d293d5d028d4 : SUCCESS in 3m 14s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/f8ac4fa0cc9d42158af8c14828a929a8 : SUCCESS in 19m 14s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/439ec98449244e9ca05a354fe023c81f : SUCCESS in 21m 30s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/b575accc1c5c4c6da07d8e582f88a446 : SUCCESS in 7m 02s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4168ae4ea5114c90983d35357aa08d42 : SUCCESS in 6m 10s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a05597795734aec9bd0a44b7ea7874f : SUCCESS in 6m 05s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/a72b9aa9b0eb4584b99fda299008d109 : SUCCESS in 7m 01s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/d36ecd5f003d4082b5c59456bad0c24f : SUCCESS in 7m 24s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/659f61be8729417395fc1b2e42849966 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":19},{"id":"7d01f1eb4fcde5ad7a052617f7f0e7b895d9db20","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-08-25 14:22:23.000000000","message":"Patch Set 19:\n\n(1 comment)\n\nLooks good to me. just small nit","accounts_in_message":[],"_revision_number":19},{"id":"5d5adb21f61ff1255b406906bc0261603928c9a3","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2020-08-26 02:05:02.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"8f32ea4f5ab672e01ac71fab64da97c9a80906ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 02:13:52.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f433f5ac454b409fa9d8e5e1f6ba248a : SUCCESS in 4m 57s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b1b5194aa1434a18a9ae87280de0a9fa : SUCCESS in 5m 04s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/7822184ade41419a96708849667bf069 : SUCCESS in 4m 23s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fd2b21274b249c685eea4b0750bb09c : SUCCESS in 4m 05s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/b9880eea2a374882bdbe03f18e3d33b4 : SUCCESS in 4m 28s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/91f18823705a4ef386b57be35a32354e : SUCCESS in 4m 25s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/7d2222972ff4483f86cd2db67ddc984d : SUCCESS in 4m 26s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/f68ed7529a354e98aedfad4a158c88b7 : SUCCESS in 4m 24s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/8b8f3b267198411380ecca752466e1b0 : SUCCESS in 6m 50s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/10c8650d0fce447eb02eb7bf159b9282 : SUCCESS in 6m 35s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/1cbf6f92053b44e980ff91f2b730ad41 : SUCCESS in 6m 39s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c52609818035439db7fe0fd1df015a06 : SUCCESS in 6m 42s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6259bcbf26545ed8a2f1b1eb79f0ad5 : SUCCESS in 6m 29s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/28d72f67a6bc42d7aaeebef24b2b1b64 : SUCCESS in 6m 41s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/3b3d2b805e9e4bcca6a60734ef43f8b6 : SUCCESS in 7m 37s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/3b55900a4e784745812cf5e55cabeea6 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":20},{"id":"cbe51f31f63e0bc7faae5373b961da493a46152b","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-08-28 11:42:18.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"ed6e4a769d7a73275e391c41b1a4dcaaf95580ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 11:42:30.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"a43d74b1ee4172aee05374a96f0eafc2694bad9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 11:57:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":20},{"id":"1e304b5b85954f7518d0ff4f0b35b5eeb0d3c319","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 11:57:30.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3c0e1ef452c483fa7ed80e287aea797 : SUCCESS in 5m 56s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/0b53fec8cbe0407ba6e637b8fd6ca796 : SUCCESS in 7m 32s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/fc4a315fb866427184f5af73916952c3 : SUCCESS in 4m 36s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6c9e2f06c144b388af2cf709dbed11b : SUCCESS in 4m 59s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/aa1e07a864e0469ca1f06c86a996bbf8 : SUCCESS in 4m 51s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/794250a8a68a490bad6eebeac7d7a816 : SUCCESS in 5m 11s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/d94123bf50624a10bf6e106c33ef851b : SUCCESS in 4m 59s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/eb1e260a5b0746c19fafa7b0a007f5ca : SUCCESS in 7m 27s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/53201a8f3e47427287010da0429422bd : SUCCESS in 7m 50s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54e758c305664a4dacfdfd5c016db74c : SUCCESS in 7m 48s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d558bfe46c3442ccb4b13910b1610011 : SUCCESS in 7m 03s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/eee5f3e726164684afb503e2e5d2d862 : SUCCESS in 6m 57s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/c26ad8a20dc74919b60837a339187cda : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"23883fa0c9a16318eb7c5dfe64587cbc98f84c3e","revisions":{"0bf86a1eaee2160b52e7a31c60f71b9f6d369353":{"kind":"REWORK","_number":1,"created":"2020-08-12 06:20:32.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/1"}}},"commit":{"parents":[{"commit":"112f9ca3bbf8bf429e5a26db6c8106a07283cc84","subject":"print completed workloads progress while running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/112f9ca3bbf8bf429e5a26db6c8106a07283cc84"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"subject":"restart the task","message":"restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0bf86a1eaee2160b52e7a31c60f71b9f6d369353"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0bf86a1eaee2160b52e7a31c60f71b9f6d369353"}]},"branch":"refs/heads/master"},"4a9754f655e3c58af484392127681ee703efe8d6":{"kind":"REWORK","_number":2,"created":"2020-08-12 06:33:12.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/2"}}},"commit":{"parents":[{"commit":"112f9ca3bbf8bf429e5a26db6c8106a07283cc84","subject":"print completed workloads progress while running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/112f9ca3bbf8bf429e5a26db6c8106a07283cc84"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:32:43.000000000","tz":480},"subject":"restart the task","message":"restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4a9754f655e3c58af484392127681ee703efe8d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4a9754f655e3c58af484392127681ee703efe8d6"}]},"branch":"refs/heads/master"},"b0d143856c84ff0a1ab9ed2aa46d02696e48c361":{"kind":"REWORK","_number":3,"created":"2020-08-13 03:03:21.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/3"}}},"commit":{"parents":[{"commit":"112f9ca3bbf8bf429e5a26db6c8106a07283cc84","subject":"print completed workloads progress while running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/112f9ca3bbf8bf429e5a26db6c8106a07283cc84"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-13 03:02:42.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b0d143856c84ff0a1ab9ed2aa46d02696e48c361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b0d143856c84ff0a1ab9ed2aa46d02696e48c361"}]},"branch":"refs/heads/master"},"686950b54ce4358413534dd8404864e035d3ccc9":{"kind":"REWORK","_number":4,"created":"2020-08-13 06:28:27.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/4"}}},"commit":{"parents":[{"commit":"112f9ca3bbf8bf429e5a26db6c8106a07283cc84","subject":"print completed workloads progress while running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/112f9ca3bbf8bf429e5a26db6c8106a07283cc84"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-13 06:28:07.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/686950b54ce4358413534dd8404864e035d3ccc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/686950b54ce4358413534dd8404864e035d3ccc9"}]},"branch":"refs/heads/master"},"2e6e6d451a926635957979053a21edc25dff4944":{"kind":"REWORK","_number":5,"created":"2020-08-13 08:59:12.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/5"}}},"commit":{"parents":[{"commit":"112f9ca3bbf8bf429e5a26db6c8106a07283cc84","subject":"print completed workloads progress while running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/112f9ca3bbf8bf429e5a26db6c8106a07283cc84"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-13 08:58:46.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2e6e6d451a926635957979053a21edc25dff4944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2e6e6d451a926635957979053a21edc25dff4944"}]},"branch":"refs/heads/master"},"b0068392f4b8a8a638b48ef078d23fa86e1d35c5":{"kind":"REWORK","_number":6,"created":"2020-08-13 10:55:00.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/6"}}},"commit":{"parents":[{"commit":"112f9ca3bbf8bf429e5a26db6c8106a07283cc84","subject":"print completed workloads progress while running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/112f9ca3bbf8bf429e5a26db6c8106a07283cc84"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-13 10:54:42.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b0068392f4b8a8a638b48ef078d23fa86e1d35c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b0068392f4b8a8a638b48ef078d23fa86e1d35c5"}]},"branch":"refs/heads/master"},"4d93c57c90622fc79760fca32f017e9f2cc55a14":{"kind":"REWORK","_number":7,"created":"2020-08-13 12:04:45.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/7"}}},"commit":{"parents":[{"commit":"112f9ca3bbf8bf429e5a26db6c8106a07283cc84","subject":"print completed workloads progress while running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/112f9ca3bbf8bf429e5a26db6c8106a07283cc84"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-13 12:03:59.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4d93c57c90622fc79760fca32f017e9f2cc55a14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4d93c57c90622fc79760fca32f017e9f2cc55a14"}]},"branch":"refs/heads/master"},"9dea0a745fbc8b80653a3225e66429f56be0f823":{"kind":"REWORK","_number":8,"created":"2020-08-14 01:17:01.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/8"}}},"commit":{"parents":[{"commit":"112f9ca3bbf8bf429e5a26db6c8106a07283cc84","subject":"print completed workloads progress while running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/112f9ca3bbf8bf429e5a26db6c8106a07283cc84"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-14 01:16:37.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9dea0a745fbc8b80653a3225e66429f56be0f823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9dea0a745fbc8b80653a3225e66429f56be0f823"}]},"branch":"refs/heads/master"},"9ef07f83467bd86f9269869058cc198ce36f9ab5":{"kind":"REWORK","_number":9,"created":"2020-08-14 02:01:27.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/9"}}},"commit":{"parents":[{"commit":"112f9ca3bbf8bf429e5a26db6c8106a07283cc84","subject":"print completed workloads progress while running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/112f9ca3bbf8bf429e5a26db6c8106a07283cc84"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-14 02:01:03.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9ef07f83467bd86f9269869058cc198ce36f9ab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9ef07f83467bd86f9269869058cc198ce36f9ab5"}]},"branch":"refs/heads/master"},"53f1e916272d61987a81a892d42ab05262c7ad80":{"kind":"REWORK","_number":10,"created":"2020-08-14 02:29:09.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/10"}}},"commit":{"parents":[{"commit":"112f9ca3bbf8bf429e5a26db6c8106a07283cc84","subject":"print completed workloads progress while running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/112f9ca3bbf8bf429e5a26db6c8106a07283cc84"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-14 02:28:53.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/53f1e916272d61987a81a892d42ab05262c7ad80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/53f1e916272d61987a81a892d42ab05262c7ad80"}]},"branch":"refs/heads/master"},"f6092703cde6998a1a9eda8a37a30f38dce6aa63":{"kind":"REWORK","_number":11,"created":"2020-08-14 03:26:11.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/11"}}},"commit":{"parents":[{"commit":"112f9ca3bbf8bf429e5a26db6c8106a07283cc84","subject":"print completed workloads progress while running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/112f9ca3bbf8bf429e5a26db6c8106a07283cc84"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-14 03:25:43.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f6092703cde6998a1a9eda8a37a30f38dce6aa63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f6092703cde6998a1a9eda8a37a30f38dce6aa63"}]},"branch":"refs/heads/master"},"5f717f60d6af481763016fa0ff654f0909c6b1e2":{"kind":"REWORK","_number":12,"created":"2020-08-14 05:50:28.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/12"}}},"commit":{"parents":[{"commit":"112f9ca3bbf8bf429e5a26db6c8106a07283cc84","subject":"print completed workloads progress while running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/112f9ca3bbf8bf429e5a26db6c8106a07283cc84"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-14 05:50:11.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5f717f60d6af481763016fa0ff654f0909c6b1e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5f717f60d6af481763016fa0ff654f0909c6b1e2"}]},"branch":"refs/heads/master"},"50dcffb3ac93f823e96c176f1f2bbde730c41a5f":{"kind":"REWORK","_number":13,"created":"2020-08-15 06:46:23.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/13"}}},"commit":{"parents":[{"commit":"b72a9866c0737f286fb1903cbb23553d6c29c22a","subject":"Merge \"Update the minversion parameter.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b72a9866c0737f286fb1903cbb23553d6c29c22a"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-15 06:45:48.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/50dcffb3ac93f823e96c176f1f2bbde730c41a5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/50dcffb3ac93f823e96c176f1f2bbde730c41a5f"}]},"branch":"refs/heads/master"},"4b59fecedbadb357a47a16a65619b2723a4f6138":{"kind":"REWORK","_number":14,"created":"2020-08-15 08:22:12.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/14"}}},"commit":{"parents":[{"commit":"b72a9866c0737f286fb1903cbb23553d6c29c22a","subject":"Merge \"Update the minversion parameter.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b72a9866c0737f286fb1903cbb23553d6c29c22a"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-15 08:21:26.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4b59fecedbadb357a47a16a65619b2723a4f6138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4b59fecedbadb357a47a16a65619b2723a4f6138"}]},"branch":"refs/heads/master"},"8d191960d082c1bbc34ea6e7963469bf5870081d":{"kind":"REWORK","_number":15,"created":"2020-08-15 09:10:59.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/15"}}},"commit":{"parents":[{"commit":"b72a9866c0737f286fb1903cbb23553d6c29c22a","subject":"Merge \"Update the minversion parameter.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b72a9866c0737f286fb1903cbb23553d6c29c22a"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-15 08:58:44.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8d191960d082c1bbc34ea6e7963469bf5870081d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8d191960d082c1bbc34ea6e7963469bf5870081d"}]},"branch":"refs/heads/master"},"1679ac4df7fd1a64fdcba3e4550e097c6d2a7524":{"kind":"REWORK","_number":16,"created":"2020-08-15 09:35:14.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/16"}}},"commit":{"parents":[{"commit":"b72a9866c0737f286fb1903cbb23553d6c29c22a","subject":"Merge \"Update the minversion parameter.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b72a9866c0737f286fb1903cbb23553d6c29c22a"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-15 09:34:53.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1679ac4df7fd1a64fdcba3e4550e097c6d2a7524"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1679ac4df7fd1a64fdcba3e4550e097c6d2a7524"}]},"branch":"refs/heads/master"},"70cacbb521bf185cfbabaf85a5b15ffd3a7d2f29":{"kind":"REWORK","_number":17,"created":"2020-08-17 01:28:10.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/17"}}},"commit":{"parents":[{"commit":"b72a9866c0737f286fb1903cbb23553d6c29c22a","subject":"Merge \"Update the minversion parameter.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b72a9866c0737f286fb1903cbb23553d6c29c22a"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-17 01:26:49.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/70cacbb521bf185cfbabaf85a5b15ffd3a7d2f29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/70cacbb521bf185cfbabaf85a5b15ffd3a7d2f29"}]},"branch":"refs/heads/master"},"7ae5b0f8e8b243bc70ebacee6ec74c4a1a510e24":{"kind":"REWORK","_number":18,"created":"2020-08-17 03:36:27.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/18"}}},"commit":{"parents":[{"commit":"b72a9866c0737f286fb1903cbb23553d6c29c22a","subject":"Merge \"Update the minversion parameter.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b72a9866c0737f286fb1903cbb23553d6c29c22a"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-17 03:36:04.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7ae5b0f8e8b243bc70ebacee6ec74c4a1a510e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7ae5b0f8e8b243bc70ebacee6ec74c4a1a510e24"}]},"branch":"refs/heads/master"},"f523aa5a1567b752febcb48c99b0c526fbe75d8b":{"kind":"REWORK","_number":19,"created":"2020-08-17 12:20:17.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/19"}}},"commit":{"parents":[{"commit":"b72a9866c0737f286fb1903cbb23553d6c29c22a","subject":"Merge \"Update the minversion parameter.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b72a9866c0737f286fb1903cbb23553d6c29c22a"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-17 12:19:45.000000000","tz":480},"subject":"[WIP]restart the task","message":"[WIP]restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f523aa5a1567b752febcb48c99b0c526fbe75d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f523aa5a1567b752febcb48c99b0c526fbe75d8b"}]},"branch":"refs/heads/master"},"23883fa0c9a16318eb7c5dfe64587cbc98f84c3e":{"kind":"REWORK","_number":20,"created":"2020-08-26 02:05:02.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/90/745790/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/745790/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/745790/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/745790/20"}}},"commit":{"parents":[{"commit":"b72a9866c0737f286fb1903cbb23553d6c29c22a","subject":"Merge \"Update the minversion parameter.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b72a9866c0737f286fb1903cbb23553d6c29c22a"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-12 06:17:44.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2020-08-26 01:56:10.000000000","tz":480},"subject":"restart the task","message":"restart the task\n\nexpect to restart one finished or aborted task, or restart some\nfailure scenarios of workloads.\n\nChange-Id: Ifebef92bc0f549e4d0c778261bfc9b40e90d39af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/23883fa0c9a16318eb7c5dfe64587cbc98f84c3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/23883fa0c9a16318eb7c5dfe64587cbc98f84c3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
