)]}'
{"id":"openstack%2Frally~770602","triplet_id":"openstack%2Frally~master~Ibd79f0c444a8cffcca0349b80aab79cd4c115158","project":"openstack/rally","branch":"master","topic":"unclosed-file","hashtags":[],"change_id":"Ibd79f0c444a8cffcca0349b80aab79cd4c115158","subject":"Fix different issues to ublock CI","status":"MERGED","created":"2021-01-13 11:37:30.000000000","updated":"2021-01-13 12:23:29.000000000","submitted":"2021-01-13 12:23:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770602-unclosed-file","meta_rev_id":"23209ac8b57cf807c0398e106c1f01c0e64426b6","_number":770602,"virtual_id_number":770602,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-13 12:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2021-01-13 11:38:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2021-01-13 11:38:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-13 11:57:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"03d2b528eef62a9ef826de887980f5331103c083","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-01-13 11:37:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3928ce12a46c6606a85923956f121030ad32a2f1","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-01-13 11:38:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37113dd2d2fb57d3f7da9e0717fbd0964fef562b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 11:57:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f43fdd61d1d143b19166a888a442cca9 : SUCCESS in 6m 38s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/f95806b1768b47d297fdeef6b41fa06b : SUCCESS in 5m 14s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/07c9515974c74b0c8ae9b97dd4b8e1a7 : SUCCESS in 4m 43s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/7050c7a6712d44058cb8a409e82ff112 : SUCCESS in 3m 50s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/fef5342c7bc34b7ea6307ca0d86f1b03 : SUCCESS in 3m 58s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/0245e54589e8436ebc7c5dbfc445bd28 : SUCCESS in 3m 51s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/36259ed88a4243cca9b87210c5d038c7 : SUCCESS in 5m 26s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/9364b0204c0f4eadaa4e4c73a143bb70 : SUCCESS in 3m 58s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/baeff3a15da346a19e9046ee1a4e8e26 : SUCCESS in 6m 54s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fc1a36d2edfd4460b070eda39ffb8cb6 : SUCCESS in 6m 40s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/fd75d85bde0e471fa4ec2d5ab7f8315d : SUCCESS in 9m 54s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e328ee12acb40fba4462d17c542089d : SUCCESS in 7m 29s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9c3d3878f634446b6c1f84cb78ef93c : SUCCESS in 7m 16s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/e3b195cd14464a46b5bcf7f130b3e704 : SUCCESS in 6m 22s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/3d368ad602264cd482a43cb765916c1c : SUCCESS in 7m 21s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/6dac635684194e058a3436f5aedcf223 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":1},{"id":"30a9fdf3f5c8622fb544b767a3ddb18fe095a6f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 11:57:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6a7079df71c35068db74af61fa87c5369386b1d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 12:23:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e127910304f94f638524f5d25f326067 : SUCCESS in 6m 02s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/da2b914726d64e209cad03689c5e97a8 : SUCCESS in 4m 20s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/6cd9313897dd4292b5117cbf705fcf97 : SUCCESS in 4m 14s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/a641086bc70643b49d827c29c47421e3 : SUCCESS in 3m 21s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/a573f16ecafe454b967fba1bb3366a0c : SUCCESS in 3m 24s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/49a3589389b74895b2f860848a8b50ae : SUCCESS in 3m 54s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/895af816396844418e0c97bbb799dc38 : SUCCESS in 3m 43s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/f566114d6a29482c9a16cb9c1e7fc92f : SUCCESS in 6m 31s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/eb21cf549424493a8c224ad3ee932640 : SUCCESS in 6m 50s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab096e100ce14604b863777fedc6842f : SUCCESS in 6m 44s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cd6e55cf59d462bb66685c38a76cf94 : SUCCESS in 6m 31s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/6504a68708a24c4bb36edad32de66561 : SUCCESS in 7m 19s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/5cfe9a4cfce64ac183059c1caeca963a : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":1},{"id":"23209ac8b57cf807c0398e106c1f01c0e64426b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 12:23:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68d035d0a575511ed505c9e64b2fdfbb1b124ca3","revisions":{"68d035d0a575511ed505c9e64b2fdfbb1b124ca3":{"kind":"REWORK","_number":1,"created":"2021-01-13 11:37:30.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/02/770602/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/02/770602/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/02/770602/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/02/770602/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/02/770602/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/02/770602/1"}}},"commit":{"parents":[{"commit":"11cf577f14e69c235eeb175f0dc4c83185728965","subject":"Propose 3.2.0 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/11cf577f14e69c235eeb175f0dc4c83185728965"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-01-13 10:50:22.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-01-13 11:30:28.000000000","tz":120},"subject":"Fix different issues to ublock CI","message":"Fix different issues to ublock CI\n\nThe recent version of unittests discovered several issues:\n\n* Context manager should be used while reading files\n  to ensure that file descriptors are closed at the right time.\n\n* wrong test data at test_consume_results_sla_failure_abort. the number\n  add_iteration execution should not be less that number of iteration\n  results.\n\nChange-Id: Ibd79f0c444a8cffcca0349b80aab79cd4c115158\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/68d035d0a575511ed505c9e64b2fdfbb1b124ca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/68d035d0a575511ed505c9e64b2fdfbb1b124ca3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
