)]}'
{"id":"openstack%2Frally~771480","triplet_id":"openstack%2Frally~master~Ie49856573dbf5bcf87837c555c046b06c569b3e9","project":"openstack/rally","branch":"master","topic":"return_code","hashtags":[],"change_id":"Ie49856573dbf5bcf87837c555c046b06c569b3e9","subject":"Let verify command return different return codes depending on results.","status":"MERGED","created":"2021-01-19 15:32:15.000000000","updated":"2021-01-22 17:52:04.000000000","submitted":"2021-01-22 17:52:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"771480-return_code","meta_rev_id":"f9e52209fe2a14a78435c04dae3e9a66ab6032b9","_number":771480,"virtual_id_number":771480,"owner":{"_account_id":27052,"name":"Marian Gasparovic","email":"marian.gasparovic@canonical.com","username":"marian"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-22 17:52:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2021-01-22 17:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2021-01-22 17:35:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-19 15:59:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-20 18:32:50.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"c3be616465baf64fbb9bda3e81081eb091b84bea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27052,"name":"Marian Gasparovic","email":"marian.gasparovic@canonical.com","username":"marian"},"date":"2021-01-19 15:32:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"126e6599b01c7c3535111fb4f4b5839281ce2947","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 15:59:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9801c75a6c3c43ec81670c9a811bb287 : SUCCESS in 5m 46s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/de16318f9b844b719b6cad0c3491f145 : SUCCESS in 4m 30s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/260d2440d089419c9dfdc260cafc1e7c : SUCCESS in 4m 41s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/4773d1a5ef984792a76da9fbd9e550bf : FAILURE in 4m 49s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/1fdd8b91ff4d4628bfcd2abbe03aa6d9 : SUCCESS in 5m 13s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/0fe69ec1de464d198d87a653bc4a06d2 : SUCCESS in 3m 52s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/29dbc666ba9f42a9a8337f5339400bb5 : SUCCESS in 5m 33s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/4e5ac9e220d74a52905a5e1eae3b5701 : SUCCESS in 4m 22s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/de68d2bd1e1447c391e986805df05af1 : SUCCESS in 7m 38s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c3274d867e034c96ae9f7540ac4452d7 : SUCCESS in 7m 00s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/10905bb04f414eceb7e706cee8d427e2 : SUCCESS in 7m 17s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0aff230e1ffe46868de88f0478054822 : SUCCESS in 7m 38s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ab64d770e214895b7e33aa6216cbb31 : SUCCESS in 6m 22s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/1307cde50b7a49bb930118e8ec7c9150 : SUCCESS in 6m 53s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/57ed49a76ed045efb167a454549b9b98 : SUCCESS in 6m 26s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/8a9671ef380c48b3953a342141ffd2c7 : SUCCESS in 8m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8660aab2759565c665db9de446c639f385c63817","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27052,"name":"Marian Gasparovic","email":"marian.gasparovic@canonical.com","username":"marian"},"date":"2021-01-19 16:46:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6db9c9ade4ca0a6f643f249fa1d1099ac1ec8d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 17:11:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ec1daa575354d5d882a852a00650f63 : SUCCESS in 5m 32s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/d657b96eef0a4360916ac68734982d7b : SUCCESS in 4m 37s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/862bbaa1e0464d6ca517740f2c626f2e : SUCCESS in 4m 39s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/97f98fba55cf443b8486e97c35d0e151 : SUCCESS in 4m 42s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/1355c7155fb24b33afd3160fda13c117 : SUCCESS in 4m 57s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/ef6a09d7623c4811be4f61455811e0c5 : SUCCESS in 4m 46s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/03ab32b7b8ff45e8889b50898f5cdc91 : SUCCESS in 4m 19s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/0184360c1d504e049ce884f754fab1b4 : SUCCESS in 4m 08s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/e2a7286cd1d3485ea6bfbdb2e5447b51 : SUCCESS in 6m 24s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/af56ea75f1814c759ffc6d6ec9b9ecb4 : SUCCESS in 6m 50s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ef7d8a19424d4218b8cf906da62958a2 : SUCCESS in 7m 05s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b32e333c8824b7f84973e6f24b2ea5f : SUCCESS in 6m 39s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a0b45233f7742ee9675b922283cc784 : SUCCESS in 6m 11s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/15382ac30e1c4ea9a83365c18453435b : SUCCESS in 7m 04s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/42cef4d48a294860b920f773cd60a437 : SUCCESS in 7m 43s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/12a011f8d3e843738cb9a32d7e3a8ee4 : SUCCESS in 8m 31s","accounts_in_message":[],"_revision_number":2},{"id":"591ad210f99526d92db59cdc7592418bf798cc51","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-01-20 18:32:50.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nIt would be nice if you can extend https://github.com/openstack/rally/blob/master/CHANGELOG.rst with [unreleased] tag and inform about changes of return code.","accounts_in_message":[],"_revision_number":2},{"id":"f26ac1e82afaeb73af7493b98fcaf84ffb9afda3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27052,"name":"Marian Gasparovic","email":"marian.gasparovic@canonical.com","username":"marian"},"date":"2021-01-22 14:37:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"48f463acc4d4ca930380be5bc05716657155300e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 14:49:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5543e25ceb448aea21412488f5448a5 : SUCCESS in 7m 50s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/cc0e4146f904499c9a65a2a746541d76 : SUCCESS in 7m 34s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/99e648c1019a4a5cab21f827cd165f33 : SUCCESS in 7m 13s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6160ce5c4bd46b38068745bdfce2fbd : SUCCESS in 4m 59s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/64a97c6c8bde418baea9722f32d6f570 : SUCCESS in 5m 46s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/bd0dc5e680f047db8305c74cec168a89 : SUCCESS in 6m 22s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/2961a57b2a7343e88ba3e9ad73b54bee : SUCCESS in 5m 45s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/9c2e6f0908714e4590e82bbd103753a6 : SUCCESS in 6m 28s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/d8697c3013f149d487ff3dfde5e44613 : SUCCESS in 7m 44s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/77bc63bb1fab4c50b125a605b2e20d86 : SUCCESS in 7m 07s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/90ca3f018fe945b8ae657d1ae1896311 : SUCCESS in 8m 54s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c6750350e434d8eb9226aa5e239e58a : SUCCESS in 6m 38s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/506101d71c554e008d0930c4eac7c53f : SUCCESS in 6m 44s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/fe49d8c9c59147c2a8b8efcd1e1f061a : SUCCESS in 7m 43s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/a10628e984fb4a47bb49d8a825bb9262 : SUCCESS in 7m 33s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/6812257d6cb646dcab83d6f72c71ba95 : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":3},{"id":"f03cf5ab5a09f86b2c0e75997424ddaf36a869fe","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-01-22 17:35:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":3},{"id":"62099ed2f78e37a534af1b810ced2a6087b86e20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 17:38:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"01cfffa70cd0d06663baa415206fa767e4bd8388","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 17:52:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a5811f58747439da2cc46b6b1178293 : SUCCESS in 7m 08s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/628fb786d9d04deab13c6b91e24328ec : SUCCESS in 6m 03s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ab37380ab7a74db7a02b5865cf3b4388 : SUCCESS in 6m 20s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/18d5728bbc9e4edb994a06c6bb57b5ea : SUCCESS in 4m 38s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/8216fd1c21f0415c9c1541863afcde0f : SUCCESS in 5m 07s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/84d1a00c763840e88d27cd9fe04a9b52 : SUCCESS in 4m 50s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/8f5ed2340a9b4feea175068a32f346c0 : SUCCESS in 5m 07s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/458d7b38b0c244beb1386ca51c6d2afd : SUCCESS in 8m 13s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/643358c158ee439cb386bdeea7460868 : SUCCESS in 7m 24s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6202887fc6944396a320b648f106ad41 : SUCCESS in 6m 44s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5995a7b1bbca494aadbd285176b519eb : SUCCESS in 6m 53s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/f9b5030c542a4a49890bdf3b668f218a : SUCCESS in 8m 35s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/4ed9ffeb45754f7a99750ee55018cbdd : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":3},{"id":"f9e52209fe2a14a78435c04dae3e9a66ab6032b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 17:52:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a72a6e1ce515e0f6720fea6a3ffbdfdea7d4e3ac","revisions":{"43eb5abfe387e81b782224869eef6d23b1812631":{"kind":"REWORK","_number":1,"created":"2021-01-19 15:32:15.000000000","uploader":{"_account_id":27052,"name":"Marian Gasparovic","email":"marian.gasparovic@canonical.com","username":"marian"},"ref":"refs/changes/80/771480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/80/771480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/80/771480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/80/771480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/80/771480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/80/771480/1"}}},"commit":{"parents":[{"commit":"412680357a5ca520812b084ee6d13650a1273cb8","subject":"Remove __unicode__() from RallyCliError","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/412680357a5ca520812b084ee6d13650a1273cb8"}]}],"author":{"name":"Marian Gasparovic","email":"marian.gasparovic@canonical.com","date":"2021-01-19 15:30:08.000000000","tz":60},"committer":{"name":"Marian Gasparovic","email":"marian.gasparovic@canonical.com","date":"2021-01-19 15:32:11.000000000","tz":60},"subject":"Let verify command return different return codes depending on results.","message":"Let verify command return different return codes depending on results.\n\nChange-Id: Ie49856573dbf5bcf87837c555c046b06c569b3e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/43eb5abfe387e81b782224869eef6d23b1812631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/43eb5abfe387e81b782224869eef6d23b1812631"}]},"branch":"refs/heads/master"},"3cc4201c7bcd6062fea734df4b16804d9ecc01ff":{"kind":"REWORK","_number":2,"created":"2021-01-19 16:46:50.000000000","uploader":{"_account_id":27052,"name":"Marian Gasparovic","email":"marian.gasparovic@canonical.com","username":"marian"},"ref":"refs/changes/80/771480/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/80/771480/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/80/771480/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/80/771480/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/80/771480/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/80/771480/2"}}},"commit":{"parents":[{"commit":"412680357a5ca520812b084ee6d13650a1273cb8","subject":"Remove __unicode__() from RallyCliError","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/412680357a5ca520812b084ee6d13650a1273cb8"}]}],"author":{"name":"Marian Gasparovic","email":"marian.gasparovic@canonical.com","date":"2021-01-19 15:30:08.000000000","tz":60},"committer":{"name":"Marian Gasparovic","email":"marian.gasparovic@canonical.com","date":"2021-01-19 16:46:03.000000000","tz":60},"subject":"Let verify command return different return codes depending on results.","message":"Let verify command return different return codes depending on results.\n\nChange-Id: Ie49856573dbf5bcf87837c555c046b06c569b3e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3cc4201c7bcd6062fea734df4b16804d9ecc01ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3cc4201c7bcd6062fea734df4b16804d9ecc01ff"}]},"branch":"refs/heads/master"},"a72a6e1ce515e0f6720fea6a3ffbdfdea7d4e3ac":{"kind":"REWORK","_number":3,"created":"2021-01-22 14:37:01.000000000","uploader":{"_account_id":27052,"name":"Marian Gasparovic","email":"marian.gasparovic@canonical.com","username":"marian"},"ref":"refs/changes/80/771480/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/80/771480/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/80/771480/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/80/771480/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/80/771480/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/80/771480/3"}}},"commit":{"parents":[{"commit":"412680357a5ca520812b084ee6d13650a1273cb8","subject":"Remove __unicode__() from RallyCliError","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/412680357a5ca520812b084ee6d13650a1273cb8"}]}],"author":{"name":"Marian Gasparovic","email":"marian.gasparovic@canonical.com","date":"2021-01-19 15:30:08.000000000","tz":60},"committer":{"name":"Marian Gasparovic","email":"marian.gasparovic@canonical.com","date":"2021-01-22 14:36:34.000000000","tz":60},"subject":"Let verify command return different return codes depending on results.","message":"Let verify command return different return codes depending on results.\n\nChange-Id: Ie49856573dbf5bcf87837c555c046b06c569b3e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a72a6e1ce515e0f6720fea6a3ffbdfdea7d4e3ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a72a6e1ce515e0f6720fea6a3ffbdfdea7d4e3ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
