)]}'
{"id":"openstack%2Frally~784918","triplet_id":"openstack%2Frally~master~I522185c079e568014822427db786e896ae41cb28","project":"openstack/rally","branch":"master","attention_set":{},"removed_from_attention_set":{"10267":{"account":{"_account_id":10267,"name":"venkata anil","email":"anilvenkata@redhat.com","username":"anil"},"last_update":"2023-10-13 13:25:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I522185c079e568014822427db786e896ae41cb28","subject":"set decorator version 4.4.2","status":"ABANDONED","created":"2021-04-06 10:45:37.000000000","updated":"2023-10-13 13:25:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0f7c05402a22ed0a07bd7909acc7baaad89ac116","_number":784918,"virtual_id_number":784918,"owner":{"_account_id":10267,"name":"venkata anil","email":"anilvenkata@redhat.com","username":"anil"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-06 10:59:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"all":[{"value":1,"date":"2021-04-08 11:38:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 10:59:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-08 11:38:14.000000000","updated_by":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"reviewer":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"state":"REVIEWER"},{"updated":"2021-04-15 15:17:25.000000000","updated_by":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"reviewer":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"state":"REVIEWER"},{"updated":"2023-09-20 14:48:30.000000000","updated_by":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"reviewer":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"state":"REMOVED"}],"messages":[{"id":"8728b013faf12b34f30f37a89a9c2d3f7af184c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10267,"name":"venkata anil","email":"anilvenkata@redhat.com","username":"anil"},"date":"2021-04-06 10:45:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62a9d070b30e3a68e6689a8be018904761eea5b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 10:59:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2832d4e8518d4e6091fa332bba4ece0b : SUCCESS in 6m 28s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/89b60819428f4798b5920c9b0195f925 : SUCCESS in 5m 45s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/465410b0298e47a9ac1100e4c4bfe657 : FAILURE in 5m 33s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/88457dea990a41b79429678385000592 : SUCCESS in 4m 53s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/4adb353ad91e4c54aad8cd447be9c684 : FAILURE in 5m 33s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/cdd16ea8864f44b2a3b92034cba28ec3 : FAILURE in 5m 54s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/a288ad385814417584c238ad6008ab0c : FAILURE in 5m 34s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/df9e8b94090342bfb0fe5df8befc8d58 : SUCCESS in 4m 24s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/b10868db107a460dbc1f1a8b2787a00b : SUCCESS in 8m 31s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a95c571e3f734e139b1c824da6d0e65d : SUCCESS in 8m 25s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/6274e4104d8f4517b33ecbd245741926 : SUCCESS in 8m 28s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e44b54cc57a74ed4a9bfa77d9c9c75e7 : SUCCESS in 7m 29s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13a2eef00f3441a5b8411f7cd14aee1f : SUCCESS in 7m 04s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/f4f449c38a4c48b48f07d4f659585789 : SUCCESS in 8m 00s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/7fad5f93d14f436bbf867c8e4434172a : SUCCESS in 7m 59s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/a9baa375cb1a46dc96426d7815134fdd : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7e32db5d3fd6d45c2c2a928f9f5b224a3b74d6f4","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2021-04-08 11:38:14.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good","accounts_in_message":[],"_revision_number":1},{"id":"c5bdffae4643ea65cdef22a58a3f96600524bee1","author":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"date":"2021-04-15 15:17:25.000000000","message":"Patch Set 1: Code-Review-1\n\nAppears decorator was now removed from rally dependencies at all: https://review.opendev.org/c/openstack/rally/+/785397","accounts_in_message":[],"_revision_number":1},{"id":"5f8b924bcfb8a72272dfde8a340ce4b074ad981a","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"date":"2023-09-20 14:48:30.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_18551\u003e with the following votes:\n\n* Code-Review-1 by \u003cGERRIT_ACCOUNT_18551\u003e\n","accounts_in_message":[{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"}],"_revision_number":1},{"id":"0f7c05402a22ed0a07bd7909acc7baaad89ac116","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-10-13 13:25:30.000000000","message":"Abandoned\n\nrally does not use decorator lib anymore","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca728bfb3c04290e5997689a66c011ecd119b2db","revisions":{"ca728bfb3c04290e5997689a66c011ecd119b2db":{"kind":"REWORK","_number":1,"created":"2021-04-06 10:45:37.000000000","uploader":{"_account_id":10267,"name":"venkata anil","email":"anilvenkata@redhat.com","username":"anil"},"ref":"refs/changes/18/784918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/18/784918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/784918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/784918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/784918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/18/784918/1"}}},"commit":{"parents":[{"commit":"a72a6e1ce515e0f6720fea6a3ffbdfdea7d4e3ac","subject":"Let verify command return different return codes depending on results.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a72a6e1ce515e0f6720fea6a3ffbdfdea7d4e3ac"}]}],"author":{"name":"venkata anil","email":"anilvenkata@redhat.com","date":"2021-04-06 10:42:04.000000000","tz":330},"committer":{"name":"venkata anil","email":"anilvenkata@redhat.com","date":"2021-04-06 10:42:07.000000000","tz":330},"subject":"set decorator version 4.4.2","message":"set decorator version 4.4.2\n\nrally commands which use decorator (i.e deployment, task) are\nfailing with traceback because of recent changes in decorator\nmodule. This patch proposes to use previous verison of decorator\nstill it gets stabilized.\n\nCloses-Bug: #1922707\nChange-Id: I522185c079e568014822427db786e896ae41cb28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ca728bfb3c04290e5997689a66c011ecd119b2db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ca728bfb3c04290e5997689a66c011ecd119b2db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
