)]}'
{"id":"openstack%2Frally~785397","triplet_id":"openstack%2Frally~master~Ia79008c6ac7dc3be92afdea5f8ffabcd697e7599","project":"openstack/rally","branch":"master","topic":"remove_decorator","hashtags":[],"change_id":"Ia79008c6ac7dc3be92afdea5f8ffabcd697e7599","subject":"Get rid of decorator lib","status":"MERGED","created":"2021-04-08 11:23:34.000000000","updated":"2021-04-12 15:05:17.000000000","submitted":"2021-04-12 15:05:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785397-remove_decorator","meta_rev_id":"bc753d5e3054c32e7e6d209b2a5adb4bf6755ebf","_number":785397,"virtual_id_number":785397,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-12 15:05:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2021-04-12 14:46:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2021-04-12 14:46:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-08 11:41:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"29a11c2fdd3f3b25e9db56e1bc696ff2e019efd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-04-08 11:23:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce7d0167cee7e957f456bee6f8bc85f89d6cf8f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 11:41:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e93e676c7bb7417294407585ff8979ec : SUCCESS in 5m 37s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/fff5f52600f04a5585286bc8597066f5 : SUCCESS in 5m 32s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/0c2bfeb7d5c343c38a72f3408f6076de : FAILURE in 4m 36s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc48bbb0681f4d8fbd690c96ba09d915 : SUCCESS in 4m 02s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/6c3646597fa940c792fccf74cdfc49e3 : FAILURE in 4m 26s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/e385920b0ab0466ab3ede2f2a353b316 : FAILURE in 4m 29s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/2034c1254fc14ebc8d6c954b65cf9738 : FAILURE in 4m 05s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/f81fe3116d844b169adb14624d1db96e : SUCCESS in 4m 39s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/864cb6538d4944b7b18124f9250b6647 : SUCCESS in 8m 14s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/910a1803ff4e4455b49b33eb4f0dace0 : SUCCESS in 7m 10s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/9f87642bfa644bc588f2c8b32fa82bb0 : SUCCESS in 7m 58s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa168aad74cf48239279387420a39b35 : SUCCESS in 6m 56s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d217fb98e0a347c89049a4e8d876ef8d : SUCCESS in 6m 21s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/a1763462cef54adf981f6ae44d1692c5 : SUCCESS in 8m 18s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/c872e67cee0243129b9740df023ebb99 : SUCCESS in 7m 02s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/0a7f41cfbee24a629786efda88e0217e : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":1},{"id":"bad9fd0e4b758e97919a25890e7c9191fe549d98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-04-12 14:15:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1c397e3da11f2db8895446025a0fe92ee094554f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 14:25:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ab88a5bfea4fe68197a1337803a45f : SUCCESS in 6m 23s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/3c863531d4164c5d95b7a734d5a7d43a : SUCCESS in 5m 27s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ec3d2a61cac84c68b03d99ed5c90483d : SUCCESS in 5m 08s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/de8d72a251c34d66953278f3965e3a37 : SUCCESS in 3m 59s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/0ac976b4efd04f948459d994fd4bec4f : SUCCESS in 4m 21s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/53c7308f37204a96847674cfbb8d10c6 : SUCCESS in 4m 38s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/0357dfaaa9104314bbf12ee6d14ca4ee : SUCCESS in 4m 28s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/5a9062502cbe4c7a893c9210cc2f9e0b : SUCCESS in 4m 24s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/0d07f6dc7172495cb9cf563a930159a2 : SUCCESS in 7m 07s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5f351087107e473bb9de3e09b15851df : SUCCESS in 9m 19s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/23ee3390e5194f169d5d3bbecb76c946 : SUCCESS in 7m 51s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6b06c04cc904d9c888ec20cb0d9883d : SUCCESS in 6m 48s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59ad4852b39b4cc1b455482bdd2aa1a2 : SUCCESS in 6m 23s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/2f99dacdf38a4e8fb357a89bedf39983 : SUCCESS in 7m 10s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/eaa858db6fc64f089d810a697ed69169 : SUCCESS in 6m 59s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/adef561e2555491f97fdda433767285d : SUCCESS in 9m 10s","accounts_in_message":[],"_revision_number":2},{"id":"a45c1965cfb513ef76e57da9041e9afe369b9e60","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-04-12 14:46:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e2019a3678fffb9b007060bc1927192dbc1e639d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 14:46:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d12aef10b5bff057115340002210732678a7457a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 15:05:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d83f59598e445382c2c815dc773f73 : SUCCESS in 5m 58s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/ccc25c13460147398e4c4cd3b110fd55 : SUCCESS in 5m 57s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/4e875be9274544c7b75f1ca150f7eaa8 : SUCCESS in 6m 12s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/32557b8841664a7e87bb38e0f3815284 : SUCCESS in 5m 02s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/a713ceb3dd374f1fa6cd182c26a2761a : SUCCESS in 4m 40s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/746bdcd3d573430b8e512c5be9706851 : SUCCESS in 5m 10s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/55438110d89a408aa6c553bb3c429446 : SUCCESS in 4m 54s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/b9a681f9e9274954ba11d56360a63e5f : SUCCESS in 7m 42s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/30a9ff26d167495485db5cc53a19c2de : SUCCESS in 7m 56s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c9e6a3b15d7424eaf15bbdf5129c153 : SUCCESS in 7m 03s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2cc6f8b3aa34ee0bc6097b584f613f5 : SUCCESS in 6m 49s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/01014eeca0df42d9858954f4b50d39a1 : SUCCESS in 7m 13s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/203be7a2eb9347a6aa71c9f54d5ec9cd : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":2},{"id":"bc753d5e3054c32e7e6d209b2a5adb4bf6755ebf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 15:05:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ee09cab5463c077f6f568d687e5f2323ccbde885","revisions":{"7ea813effc98545b7e53a7d25fcf10d9c99a0370":{"kind":"REWORK","_number":1,"created":"2021-04-08 11:23:34.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/97/785397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/97/785397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/785397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/785397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/785397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/97/785397/1"}}},"commit":{"parents":[{"commit":"a72a6e1ce515e0f6720fea6a3ffbdfdea7d4e3ac","subject":"Let verify command return different return codes depending on results.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a72a6e1ce515e0f6720fea6a3ffbdfdea7d4e3ac"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-04-08 11:21:12.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-04-08 11:21:12.000000000","tz":180},"subject":"Get rid of decorator lib","message":"Get rid of decorator lib\n\nThis lib is used only in two places and the latest release fails now.\n\nChange-Id: Ia79008c6ac7dc3be92afdea5f8ffabcd697e7599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7ea813effc98545b7e53a7d25fcf10d9c99a0370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7ea813effc98545b7e53a7d25fcf10d9c99a0370"}]},"branch":"refs/heads/master"},"ee09cab5463c077f6f568d687e5f2323ccbde885":{"kind":"REWORK","_number":2,"created":"2021-04-12 14:15:27.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/97/785397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/97/785397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/785397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/785397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/785397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/97/785397/2"}}},"commit":{"parents":[{"commit":"a72a6e1ce515e0f6720fea6a3ffbdfdea7d4e3ac","subject":"Let verify command return different return codes depending on results.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a72a6e1ce515e0f6720fea6a3ffbdfdea7d4e3ac"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-04-08 11:21:12.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-04-12 14:15:09.000000000","tz":180},"subject":"Get rid of decorator lib","message":"Get rid of decorator lib\n\nThis lib is used only in two places and the latest release fails now.\n\nChange-Id: Ia79008c6ac7dc3be92afdea5f8ffabcd697e7599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ee09cab5463c077f6f568d687e5f2323ccbde885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ee09cab5463c077f6f568d687e5f2323ccbde885"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
