)]}'
{"id":"openstack%2Frally~787117","triplet_id":"openstack%2Frally~master~I6ff751c558098f1c3210fd9496c2cfceaf2092c1","project":"openstack/rally","branch":"master","hashtags":[],"change_id":"I6ff751c558098f1c3210fd9496c2cfceaf2092c1","subject":"Use py3 as the default runtime for tox","status":"NEW","created":"2021-04-20 06:54:50.000000000","updated":"2021-04-20 07:13:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f0df961696d250316d627106ee766f3982faf2e6","_number":787117,"virtual_id_number":787117,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-20 07:13:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 07:13:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee62bab99c87ad6e8c79004b27d484e1abaa300c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2021-04-20 06:54:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0df961696d250316d627106ee766f3982faf2e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 07:13:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a851165d2ed4201b01c63de2d33980a : SUCCESS in 5m 36s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/adc1d67a85c44f6e9e94a76b21685417 : SUCCESS in 9m 39s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ea00ab4643f544f69eb5da2393b23ace : SUCCESS in 5m 54s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3c16c516a624bb7ba48b59917f021ff : SUCCESS in 4m 41s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/5f112c8398714b6093d60dd2846fcb91 : SUCCESS in 6m 07s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/bc7ce2bf1ab34230adef828926f124d1 : SUCCESS in 5m 34s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/1715549db1424e568b8af6f894727c45 : SUCCESS in 5m 00s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/a9d4378332df44479bc7f1118d3e2898 : SUCCESS in 5m 15s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/2b9ffddba3a84ce4841b48dd71f4c4dd : SUCCESS in 6m 50s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c6603767dc4341f1a2fceabd8e8802bd : SUCCESS in 9m 22s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/a43d3608637344babebc26c8ac408d53 : SUCCESS in 8m 34s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e2840d3ad3b49eaa6b8da7e3c9d2403 : SUCCESS in 7m 41s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1775017465d04bb9956d5b25e392b29c : SUCCESS in 7m 08s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/4f39899816e04b1c942c42f95934b309 : SUCCESS in 12m 17s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/3620dffd7a6643efb4127563d9bb70b7 : SUCCESS in 8m 37s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/059ab025eeb84e73b5ebdea20d3b20dc : SUCCESS in 9m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b55404ae6e0130f1d7b2aec8d26a8c0acacbda32","revisions":{"b55404ae6e0130f1d7b2aec8d26a8c0acacbda32":{"kind":"REWORK","_number":1,"created":"2021-04-20 06:54:50.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/17/787117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/17/787117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/17/787117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/17/787117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/17/787117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/17/787117/1"}}},"commit":{"parents":[{"commit":"ee09cab5463c077f6f568d687e5f2323ccbde885","subject":"Get rid of decorator lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ee09cab5463c077f6f568d687e5f2323ccbde885"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 06:53:35.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 06:54:44.000000000","tz":480},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: I6ff751c558098f1c3210fd9496c2cfceaf2092c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b55404ae6e0130f1d7b2aec8d26a8c0acacbda32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b55404ae6e0130f1d7b2aec8d26a8c0acacbda32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
