)]}'
{"id":"openstack%2Frally~789465","triplet_id":"openstack%2Frally~master~I7e940ea0dce10985bdf8b67deeb3a4bed86523b5","project":"openstack/rally","branch":"master","hashtags":[],"change_id":"I7e940ea0dce10985bdf8b67deeb3a4bed86523b5","subject":"setup.cfg: Replace dashes with underscores","status":"ABANDONED","created":"2021-05-04 02:22:25.000000000","updated":"2021-06-16 08:42:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c9add49127bb25a1b3f2f7981adcdb0f84427404","_number":789465,"virtual_id_number":789465,"owner":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":0,"date":"2021-06-16 08:37:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2021-06-16 08:42:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2021-06-16 08:42:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-04 02:36:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-16 08:37:20.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"e8c7a6bddfdca21809a369a80664d9f43a97cd60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2021-05-04 02:22:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f732c6c6fb78261665fd7ca76da6cdbe1a86e772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 02:36:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1400271d78db4964a8cca8b77cf671e8 : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f143acee72a7466f800d600d8bbe1642 : SUCCESS in 7m 37s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b4347e124f18453e9b05aa0443256457 : SUCCESS in 5m 04s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f15fba8713aa49eca6e91823e50f46ee : SUCCESS in 6m 29s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/15f03bef4e3a4b85bc73bae363e9969f : SUCCESS in 4m 07s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/a6d54c116b074eb6a891f637a3da3539 : SUCCESS in 4m 43s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/b3d1328fc69e469d9bcbb9e4b61e5617 : SUCCESS in 5m 25s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/6563ccdfaded4275b6a3c1f102e12bb8 : SUCCESS in 5m 06s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/57a4b55a79b843468eb1112db74518d4 : SUCCESS in 3m 46s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/3940e9d8c72c4341892b79e5094051d8 : SUCCESS in 7m 03s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bec1c9c8558d4bc98f792cfb7ce4b887 : SUCCESS in 7m 00s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/37ba39f1a73249919cb035c2921a925f : SUCCESS in 7m 53s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57e5e05bc7f949a297c1212c7efc5c86 : SUCCESS in 6m 29s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2dc2f5618c534f2ab122f5eae4940bf1 : SUCCESS in 7m 19s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/f3d2499e581f4664b31556ca15033476 : SUCCESS in 7m 42s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/8026ccbfe1734a108e20a93cd1257a08 : SUCCESS in 6m 48s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/bf6918e59dc34adfb90db6d231b10a40 : SUCCESS in 11m 29s","accounts_in_message":[],"_revision_number":1},{"id":"60c4e804117294f78d13910966376637cb3e0676","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 08:37:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"96f740c68222d8aabc8c12c506168ec0716e6959","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 08:37:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4212888af1c674589e8bc03dc21750cec837d9b8","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 08:42:14.000000000","message":"Patch Set 1: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"c9add49127bb25a1b3f2f7981adcdb0f84427404","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 08:42:40.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/rally/+/789039/ was proposed first","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"246bcd88d788323f80dfaf87f2d715730cf16c60","revisions":{"246bcd88d788323f80dfaf87f2d715730cf16c60":{"kind":"REWORK","_number":1,"created":"2021-05-04 02:22:25.000000000","uploader":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"ref":"refs/changes/65/789465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/65/789465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/65/789465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/65/789465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/65/789465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/65/789465/1"}}},"commit":{"parents":[{"commit":"ee09cab5463c077f6f568d687e5f2323ccbde885","subject":"Get rid of decorator lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ee09cab5463c077f6f568d687e5f2323ccbde885"}]}],"author":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2021-05-04 02:20:47.000000000","tz":480},"committer":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2021-05-04 02:22:20.000000000","tz":480},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: I7e940ea0dce10985bdf8b67deeb3a4bed86523b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/246bcd88d788323f80dfaf87f2d715730cf16c60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/246bcd88d788323f80dfaf87f2d715730cf16c60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
