)]}'
{"id":"openstack%2Frally~792311","triplet_id":"openstack%2Frally~master~Ia5702adb636350b45aea279f79ebd078c07b9491","project":"openstack/rally","branch":"master","topic":"fix-config","hashtags":[],"change_id":"Ia5702adb636350b45aea279f79ebd078c07b9491","subject":"Fix invalid interpolation syntax","status":"NEW","created":"2021-05-20 01:46:09.000000000","updated":"2021-05-20 03:44:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e46d1591fe82a05b22e6b4f00225b27825fb3fff","_number":792311,"virtual_id_number":792311,"owner":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-20 03:44:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-20 02:13:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad4893b7b404951143124edbe3bc61cd6458b2e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"date":"2021-05-20 01:46:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"606826285e6369e3f7e833ebc0d1ee7f7df3d8ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 02:13:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0963c1cf6cc840b4ac6315ecb62f6fd3 : SUCCESS in 4m 48s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e5250381acc5453ab4958483c5890a40 : SUCCESS in 5m 11s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/170d651770734677b06c4f6a667cb7f0 : SUCCESS in 5m 13s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/20e048fb2968494596cedde325c81a2b : SUCCESS in 6m 01s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/b6b3b865989746bdb075a88c397df42d : FAILURE in 3m 52s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/8ba4a2710f0743a8a1d88c188114a17d : FAILURE in 4m 34s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/50edd17fa6b845a5a387d9b716d2f57d : FAILURE in 4m 25s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/9e803c12c0564850a7afa77ac0d303a2 : SUCCESS in 3m 58s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/bcf55b67d6cf46549676f5677a8cc471 : SUCCESS in 7m 26s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e0293c2866874f0c856b210aa446cf6c : SUCCESS in 7m 25s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ad26e323abaa4e759b873f9eb461b34c : SUCCESS in 8m 47s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/699c26c77871489fb752e0a741cdf6d4 : SUCCESS in 7m 06s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8e410743e734fbb88ba05f065ecd642 : SUCCESS in 6m 28s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/22849185209a4d95988d321fb4a08698 : SUCCESS in 7m 07s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/534f31b84da941b3b2cdbb42dcf7dad4 : SUCCESS in 6m 44s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/b416f282fb354652a10335404aa3808e : SUCCESS in 9m 40s","accounts_in_message":[],"_revision_number":1},{"id":"31c58fa9d4b2c54eb72255141bd6bba49e36c63c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"date":"2021-05-20 03:25:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"83749244bf7e7cf6b51f68b8a3fa35b085744319","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"date":"2021-05-20 03:32:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e46d1591fe82a05b22e6b4f00225b27825fb3fff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 03:44:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1781d912e604402c9f6c2dc097495196 : SUCCESS in 4m 30s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/15bb5cf4749d44ffbbd87d20368d4bee : SUCCESS in 7m 44s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/54a71edf10db4c3a8ececa3c8077ab59 : SUCCESS in 5m 29s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff6d7265908e491ba6427022dc306cc7 : SUCCESS in 3m 36s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/f41a79a6aba74719a5287f7a043950af : SUCCESS in 4m 00s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/7a79cc7756284c23ba7ed1ab634c9b78 : SUCCESS in 4m 25s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/930420052afa4273b717a382e07a7e36 : SUCCESS in 4m 31s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/9d5a2a4b33124749a08f8672ded323fe : SUCCESS in 4m 28s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/19b4066d2bc34e28959e8c2884d475d7 : SUCCESS in 6m 05s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7e3f3338811649de91ce46b70804122f : SUCCESS in 7m 07s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/b543ccbb263c4529a1989e12df10eb42 : SUCCESS in 7m 19s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4aa4831e62840cfbd8a7d574e4785ff : SUCCESS in 6m 37s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/340f1e049db14bf3b398eab196bb926b : SUCCESS in 7m 18s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/57a29b34442b49baa6822e7483e9ee31 : SUCCESS in 7m 21s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/11c15ede514840b5b5ab2d4151cbfc4d : SUCCESS in 6m 45s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/03a42b40146a492dbeba13f4875f1b98 : SUCCESS in 8m 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"93a98b4f61307daa1050ccdd19fab9f5c445bb23","revisions":{"03f72dc3b1846fdffd93252e0e722e2a8818108a":{"kind":"REWORK","_number":1,"created":"2021-05-20 01:46:09.000000000","uploader":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"ref":"refs/changes/11/792311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/11/792311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/11/792311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/11/792311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/11/792311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/11/792311/1"}}},"commit":{"parents":[{"commit":"8a1c326137e4a3cb47f84ce0255098cec31e4943","subject":"Keep using bionic for tox jobs except with Python 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8a1c326137e4a3cb47f84ce0255098cec31e4943"}]}],"author":{"name":"ZijianGuo","email":"guozijn@gmail.com","date":"2021-05-20 01:41:17.000000000","tz":480},"committer":{"name":"ZijianGuo","email":"guozijn@gmail.com","date":"2021-05-20 01:46:06.000000000","tz":480},"subject":"Fix invalid interpolation syntax","message":"Fix invalid interpolation syntax\n\nIf we use the configuration that contains %s in the extra options,\nsuch an error will occur. Therefore, RawConfigParser should be used.\n\nSigned-off-by: ZijianGuo \u003cguozijn@gmail.com\u003e\nChange-Id: Ia5702adb636350b45aea279f79ebd078c07b9491\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/03f72dc3b1846fdffd93252e0e722e2a8818108a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/03f72dc3b1846fdffd93252e0e722e2a8818108a"}]},"branch":"refs/heads/master"},"92853d78fa64a889b0febd713fb70def9e81e2e4":{"kind":"REWORK","_number":2,"created":"2021-05-20 03:25:57.000000000","uploader":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"ref":"refs/changes/11/792311/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/11/792311/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/11/792311/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/11/792311/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/11/792311/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/11/792311/2"}}},"commit":{"parents":[{"commit":"8a1c326137e4a3cb47f84ce0255098cec31e4943","subject":"Keep using bionic for tox jobs except with Python 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8a1c326137e4a3cb47f84ce0255098cec31e4943"}]}],"author":{"name":"ZijianGuo","email":"guozijn@gmail.com","date":"2021-05-20 01:41:17.000000000","tz":480},"committer":{"name":"ZijianGuo","email":"guozijn@gmail.com","date":"2021-05-20 03:25:47.000000000","tz":480},"subject":"Fix invalid interpolation syntax","message":"Fix invalid interpolation syntax\n\nIf we use the configuration that contains %s in the extra options,\nsuch an error will occur. Therefore, RawConfigParser should be used.\n\nSigned-off-by: ZijianGuo \u003cguozijn@gmail.com\u003e\nChange-Id: Ia5702adb636350b45aea279f79ebd078c07b9491\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/92853d78fa64a889b0febd713fb70def9e81e2e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/92853d78fa64a889b0febd713fb70def9e81e2e4"}]},"branch":"refs/heads/master"},"93a98b4f61307daa1050ccdd19fab9f5c445bb23":{"kind":"REWORK","_number":3,"created":"2021-05-20 03:32:58.000000000","uploader":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"ref":"refs/changes/11/792311/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/11/792311/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/11/792311/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/11/792311/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/11/792311/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/11/792311/3"}}},"commit":{"parents":[{"commit":"8a1c326137e4a3cb47f84ce0255098cec31e4943","subject":"Keep using bionic for tox jobs except with Python 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8a1c326137e4a3cb47f84ce0255098cec31e4943"}]}],"author":{"name":"ZijianGuo","email":"guozijn@gmail.com","date":"2021-05-20 01:41:17.000000000","tz":480},"committer":{"name":"ZijianGuo","email":"guozijn@gmail.com","date":"2021-05-20 03:32:47.000000000","tz":480},"subject":"Fix invalid interpolation syntax","message":"Fix invalid interpolation syntax\n\nIf we use the configuration that contains %s in the extra options,\nsuch an error will occur. Therefore, RawConfigParser should be used.\n\nSigned-off-by: ZijianGuo \u003cguozijn@gmail.com\u003e\nChange-Id: Ia5702adb636350b45aea279f79ebd078c07b9491\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/93a98b4f61307daa1050ccdd19fab9f5c445bb23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/93a98b4f61307daa1050ccdd19fab9f5c445bb23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
