)]}'
{"id":"openstack%2Frally~796454","triplet_id":"openstack%2Frally~master~I96666afc1ad9a5c64ca5b98647cb0ba185389524","project":"openstack/rally","branch":"master","topic":"uc","hashtags":[],"change_id":"I96666afc1ad9a5c64ca5b98647cb0ba185389524","subject":"Update requirements","status":"MERGED","created":"2021-06-15 12:21:53.000000000","updated":"2021-06-16 10:01:36.000000000","submitted":"2021-06-16 10:01:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796454-uc","meta_rev_id":"1f2e7ce93a9b8147e1b828e9740f361a12e5b931","_number":796454,"virtual_id_number":796454,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 10:01:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2021-06-16 09:24:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2021-06-16 09:24:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 12:43:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09f722232d5f3eac635b93922d798c22e40eb4bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 12:21:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c51f764f11c5a38732fbd2225c0b45627f9e70d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 12:43:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/560ccd9be0cc41cdb9795c0d2bce57f6 : SUCCESS in 7m 37s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/0dcded82cc9548d899f123560d788f01 : SUCCESS in 5m 36s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/1e3d326d7d8141a7b8692a5006078faf : SUCCESS in 7m 31s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a55baebbd448fdb0b75a18b3c08977 : SUCCESS in 4m 20s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/ccbc8ba251f54ef5b54fc7bb41e0a55c : FAILURE in 4m 22s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/e36f25a99bf3427098ad97f1d5a7b34f : FAILURE in 5m 41s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/8e5dc8e67606470a838130c7954e4ee9 : FAILURE in 5m 45s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/b39ccf8e4a69425897ac14091fdd7ffe : SUCCESS in 5m 02s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ef7c852bd70b4e2291a1ce034b4e69bc : SUCCESS in 8m 43s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/11625855e091405c9a8bea42e18680b8 : SUCCESS in 7m 02s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/f2eef259ab604dc58ccbd95a07594de7 : SUCCESS in 7m 52s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/581f8fe138334a84ab553722b944b112 : SUCCESS in 7m 28s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e82f7f81032b4f69bb3e80be2f549181 : SUCCESS in 7m 03s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/f44a8a4d7f9a4dacb384ae573ed9312e : FAILURE in 4m 29s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/1e94e7032db54dd5a24e235916222921 : FAILURE in 3m 43s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/944be0abdab24544974851527f31a6f4 : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a3a37da57c1fbba3c68e4143ff4ba9faf605b25e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 12:51:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"52f312d4e9f87658d450c7e3ac07719c8d2d67b8","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 12:51:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ae4cad19552b460f3c571bdce6959905161e34ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 13:09:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/93dd17fc5c034417aa7056e181879c10 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b79614f84d634b7f8a650a6181bc9622 : SUCCESS in 5m 58s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/382e8898dcf64b28ae4b035943e4706c : SUCCESS in 5m 29s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/55dcad934a42430cb1f8bd013454375d : SUCCESS in 4m 57s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/056f1155899f403d95156256adf425e3 : SUCCESS in 4m 17s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/ba6ba0929afa4b03ab319732276cf180 : SUCCESS in 4m 35s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/b4d9aad24dad4db58a98d2d2402d5d52 : SUCCESS in 4m 41s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/755d9c6529be468e83f5af5996596a30 : SUCCESS in 4m 09s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/2cf566771d2a473094c2a19651965461 : SUCCESS in 4m 38s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/c14f3ffb7a3b4706ae815f694d0feef7 : SUCCESS in 7m 15s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3367e896b6734912bf813211c8d78f04 : SUCCESS in 7m 42s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/dbfe0cee7f3443ea9e8f46fd3c33b2dd : SUCCESS in 7m 32s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9784bbac49b44a0fa0dd475beb49ff53 : SUCCESS in 7m 04s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb822ceaeb5c4a358a5e71a3e1e76f0a : SUCCESS in 6m 53s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/19c5fd643037477095eb9c35c542cf2e : FAILURE in 3m 18s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/fc25afe7c6e54c1390b6a2d9e68fa48e : FAILURE in 3m 38s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/bda422889dfb4d9ab549e6a918ff3b76 : SUCCESS in 9m 16s","accounts_in_message":[],"_revision_number":2},{"id":"c304cd61681c43410c42ef7d59ee1f68025b3d4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 13:50:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"af0584d0acaf249d69680a4f78b571e799de3668","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 13:50:48.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f2d02a358b7dcb3d35769a170fbf6bbbea20a570","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 13:51:56.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"05d097cd93d39d5e87b0c59ae6df803b5823fcd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 13:54:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ac358e5fd3a9364365f8ebaf5cf5a16c105b96fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 14:36:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a3c63b9572d649e19c4899135c2bdf51 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62cb7e685ce34b618d378f0dab61fab9 : SUCCESS in 7m 21s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/47bcd44b0b7c4763acc9ffb159d4d2f5 : SUCCESS in 5m 45s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f4e67e293c4e4702b7411c43be54b016 : SUCCESS in 5m 58s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/d439d4652ac84fbca2360a4b56d4eb43 : SUCCESS in 5m 18s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/0c48d64b73dc478a90dcbb11bf36720b : SUCCESS in 5m 36s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/78052fe2d29848b9a09446314eeb668e : SUCCESS in 5m 58s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/e2a44e107c414421aaa9ce8edeee6f45 : SUCCESS in 6m 17s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/e30b66ca05b7407da312df70801b5741 : SUCCESS in 5m 12s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/35334d35fa0a4ee8908b3c29c20a6136 : SUCCESS in 7m 27s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fbb6fb7dcfd54229811fdd0cf259276a : SUCCESS in 7m 24s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/e2c68a3cb47247029ce76059519508ee : SUCCESS in 8m 10s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ea19b28394848b6bca3256b843cb0f7 : RETRY_LIMIT in 2m 37s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/559f576aef9b4f03b4ec27974638ebc2 : RETRY_LIMIT in 2m 35s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/862d8b771758488fa69bd897d26d1219 : RETRY_LIMIT in 3m 35s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/c86201369b6b420ea13d150ec32a3490 : RETRY_LIMIT in 3m 17s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/1d06021cce7c40bc82fbfef738ba4b7a : SUCCESS in 11m 52s","accounts_in_message":[],"_revision_number":5},{"id":"b7d89063ed3357e9c05909842fe72c11701c7acd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 15:02:24.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"cb0aaf7dbab4962a3421cf2d9629b2ad10f7f443","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:35:59.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fb10eeb479dc4626b7ce05ec8c7cbddb : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1c607bd0d3a41d7a9cce3ca602cd8ff : FAILURE in 17m 29s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/1022c72c37034e3c8dc0dcc42c4efc64 : FAILURE in 18m 45s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/c65192ce7bec4ce0893b8cf3307975a6 : FAILURE in 16m 34s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ca6ce78c03d4813a09d77782a94448a : FAILURE in 6m 09s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/96f54192eabb47938d19b578bce1e598 : SUCCESS in 5m 24s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/f66eae31e5e047258115bb4000b70095 : SUCCESS in 4m 51s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/dab3d92033a44a9c8ed28d48c3242885 : SUCCESS in 5m 02s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/fb679f18b744405ea4ab5f26e934c4ff : SUCCESS in 4m 41s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/c0a82d5d9bab4eccb655c41cb553e89f : SUCCESS in 8m 41s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/df0d1d9994cf4c5da0c12fb198f3200a : SUCCESS in 8m 28s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/4dda497eb2cf473ab552a2b870daab2c : SUCCESS in 9m 06s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5da079dcd7e443e88088241386ea4a90 : RETRY_LIMIT in 3m 48s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7da25a65813f440abd79dba1f51e3f3a : RETRY_LIMIT in 2m 47s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/3b6dfc7eb5b945a59974d5969da6efc3 : RETRY_LIMIT in 3m 56s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/04881310201f4adf8f266d55f5a40aec : FAILURE in 4m 22s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/ac4ad03295524cc09060221cbb3e9b35 : SUCCESS in 11m 33s","accounts_in_message":[],"_revision_number":6},{"id":"c328a79cc6ae4949351894207198e02de4e66c75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 15:41:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7e9a5acd03c483bdf14ad5b2fbe6f1b393f92156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 15:55:49.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d8d37a36d9a22535ca0491094cdd3dc668f2a7de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 16:12:15.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"44ab768ccf6929805a41ad3aca843c539adfd34d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 16:27:42.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7610a6645585852181db004efe1174a669d6e8cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 16:50:04.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"e86cf091ff94f48963b6dbf8855c614ebe303a91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 17:06:24.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1ff0feeab64e4dd8a6c823bb59f15cb0 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88421ac4049d48e3b769bee0fd3118fe : SUCCESS in 6m 28s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2462fdbec0b8445ba7ca879a08b3d01e : SUCCESS in 9m 29s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83d20876e2314e14ae95021d9c167bcf : SUCCESS in 7m 06s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/e9bd8d6258d44c77929d555dbbf953ef : SUCCESS in 9m 00s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/3fbe550e7d364ca19b8fcfcf193a8d96 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":11},{"id":"3794dd89ae3823171ff6bdb8a28b9ca0afd31fda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 17:08:36.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"c76b029ee76aefe1c8a14f6bb636b0ecceeac720","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 17:09:33.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a8c572c33724db46dd73cc1928c0bde12cfd8df1","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 17:09:42.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"93cefa577124a51612f4692daf199992391091e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 17:32:08.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d97334b4abed4d3cbc4b370919a5b11a : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f368c07446642c4b123ee5726d8f115 : SUCCESS in 5m 53s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/0b210616e8f44fef9ec836279df7aacb : SUCCESS in 5m 33s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/7f15827ae55842f087bca8e1a3aa3350 : SUCCESS in 5m 34s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f5ca68b7bc437c9f729b4d3cb39c6f : SUCCESS in 4m 11s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/e2f70d87f6cc4c5c8f0586b501ed0a39 : SUCCESS in 4m 13s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/b3496be894164618a599e0b50adb4a76 : FAILURE in 3m 52s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/b16e5c21e288486292ec7a75a20a63f4 : RETRY_LIMIT in 3m 16s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/eecf9065eef449ea9a9f1d832860ec1c : SUCCESS in 4m 08s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/302c2db69eb8485ab322d4063bf257c4 : SUCCESS in 7m 53s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/68d5b3b8111646509f270fd5e49f9ea8 : SUCCESS in 7m 05s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/379a3e0ec9144fb0bbc53e60db5f2306 : SUCCESS in 7m 52s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a56fd13b0df6425bb522fa792bbfeacf : SUCCESS in 7m 07s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2628c0f91d404091b6d5fb84a43eb188 : FAILURE in 5m 56s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/0497981974cb4150b8a127a2cd7fb951 : SUCCESS in 8m 50s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/a00ae43880184172a6ab4e17026eadc9 : FAILURE in 4m 45s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/f80538e7f0bb44448ae79b1d0cb9a2aa : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":13},{"id":"1c59614660f04376be87dc8c2344f6d4f87d5994","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 17:59:43.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"33c14967a195081cf579fb2fe8aa1bf04ec1aba8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 18:19:44.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/24f7262a55fa4fbb8ba3f7caeab043f5 : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/274ea3727374469d8935f39b0f411b6e : SUCCESS in 8m 19s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e3493759f706485c8fc713f111214b0b : SUCCESS in 5m 47s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/b3e3e2bb08f242cd9f1f39053153f68a : SUCCESS in 5m 36s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d57271927c2400b847a50e75dad116d : SUCCESS in 4m 15s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/985f6519fd7f49d2bf28bb2451558a52 : SUCCESS in 4m 48s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/e2cce71de07a464aa70416d77a74c9cd : FAILURE in 3m 31s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/b9533b423c1b422d80fe762d5564cfa5 : RETRY_LIMIT in 3m 12s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/8a1e49f710844a82b5dc61fe7ac5ac5f : SUCCESS in 4m 07s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/92b0a24c89ef4766a687c57c6718acea : SUCCESS in 6m 56s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a0355bc1b5cb452ea90d25a17d30fc2f : SUCCESS in 6m 48s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/102a9aaa9b3542fbaa73726430d293bc : SUCCESS in 7m 40s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5170b8236b1e4c4da0207f2d51788302 : SUCCESS in 6m 38s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c70b0ff6ed3c4c86a89f0753ee4d845d : SUCCESS in 6m 47s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/b6efd36391c3463f8932034757d777d4 : SUCCESS in 8m 34s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/fa929ff731ec4c1cbadfe5d37186d696 : SUCCESS in 7m 30s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/4ddfe8ee734140bb8ea567571da451b6 : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":14},{"id":"ee2fb6beabb982f80396da96b342f471cab8146c","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-15 19:50:29.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"4c5f342b151e8a07f1068857c5969fe73b629c5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 20:03:59.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b0099b2078834c158429e1d214416aa9 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b83df9bfa44c4c058767d1f0ff534705 : SUCCESS in 5m 30s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/ecc13fc684ca4ef287385d87170ca77a : SUCCESS in 5m 17s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/926da85471f34c53b78596e140ae9ba3 : SUCCESS in 5m 21s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcdc2bc20dcd493f8cf874a45da8ca11 : SUCCESS in 4m 46s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/aa6ed8e14ff349249f97fe7b05c77aca : SUCCESS in 5m 07s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/31c3e75335684a1b82145b8caf870a08 : FAILURE in 3m 47s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/d9fc61b63b574509a7dc5ced1121019f : RETRY_LIMIT in 3m 12s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/6c412f8cc62046e48548f25aaf49fab9 : SUCCESS in 4m 13s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/0f3dab465f27480597172a24b5264abd : SUCCESS in 7m 15s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/333d10f68fd84187928bfaeb8aff23ae : SUCCESS in 7m 16s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/250008f4741a46b39892dccbe108ac65 : SUCCESS in 7m 43s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0506e50d357e414f8393960f84b6d742 : SUCCESS in 6m 48s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a3c8bf835234b729eb5bf90cd533152 : SUCCESS in 7m 07s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/71744c31850845ce83a49f542445d159 : SUCCESS in 8m 17s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/3969dc8bb19f46729f48c151b051dbf4 : SUCCESS in 7m 17s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/b43a32560c5d4da3b4840b28dc18d84c : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":14},{"id":"84e725569e3936474e416eb490e6a7061b7fa3b1","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 08:13:46.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"4d0ad7c71af387f062efbd75bb2d9da6d45bb792","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 08:35:29.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"8b0e3f36e398def893f204c6c2ee714007481383","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 08:54:41.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1be0962fbf864fb3b142e2d373086279 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/907a211186614cf3b0994c6b2ca601de : SUCCESS in 5m 46s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/99de853c2e0c47cb9b329c3d1c9f8033 : SUCCESS in 5m 26s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f6e377e4099d4f29a135107f829b29cb : SUCCESS in 6m 47s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/db83b10f3e9343a0992b32a33cc22650 : SUCCESS in 5m 46s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/7857e32f28bc4f32856b4892c8edcef9 : SUCCESS in 5m 28s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/a01a4903eedc419e963fa45c4b639cb3 : SUCCESS in 5m 35s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/33e05886b57d43dcaff39f6ccbbf1372 : SUCCESS in 4m 30s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/a732927a7fba4e99a905f0b2d740eae8 : SUCCESS in 4m 20s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/9e2d136196554b82aeda2035726ba5a4 : SUCCESS in 7m 59s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4fac50d9a7374964bb5f0b413aeba460 : SUCCESS in 7m 33s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/5df4dcb404e5460d89a9661cfbb4cc2a : SUCCESS in 7m 35s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e9a6ec8eb9c4cd6877b460298bf45b5 : SUCCESS in 7m 29s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/233eaa0e3d8d49ffb2a65bceacd3116c : SUCCESS in 7m 32s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/3f9e737ae7db44c8a780605477f08a0c : SUCCESS in 9m 04s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/7747e7c2aa8b4c17bd24ab213953f9b0 : SUCCESS in 7m 52s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/cae84848d97e4df5896a43aad4b6b4b8 : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":15},{"id":"c6243e79bc49e16890b4f2f0b385399bc53a714c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 09:01:25.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"8ba2c4fe84c8baf2dd0603a3587431f4ccc91c42","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 09:01:57.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"91fd3f2fce1437e41f13ecf5f969f2a50c168166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 09:24:43.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"205d85a3678265129570056170fd523d91616196","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 09:24:55.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"f8e27cd6391a04e1103dc292185d13e5fc9df2a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 09:46:58.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7e1e9ee4483d4271a1caf67f4071683d : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0910428b814347b59ed07c1e373418 : SUCCESS in 7m 45s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/f1903c73b04a4fd1b7517f5a63b7437f : SUCCESS in 5m 22s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/046064367270494a9f53f3f84e16610e : SUCCESS in 5m 14s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa9188c967c84b2aa24b2f4c06a2637d : SUCCESS in 5m 08s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/2f877652239a4187bedbb45fca16e2b2 : SUCCESS in 5m 06s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/0687fcd73f224d92814c8cd7ff76cc7b : SUCCESS in 4m 41s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/76b39a6eef2b4bfaa472bb241803b6b2 : SUCCESS in 5m 06s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/ad0daa8923c34940a81707d0d7ce2442 : SUCCESS in 4m 01s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ba543b0274464bb6b8332592fb548619 : SUCCESS in 7m 03s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/71c1131894824a9f95ca395289fb89eb : SUCCESS in 9m 37s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ea3f9b6de28b455ea93ecd599c2151bb : SUCCESS in 7m 27s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1515458d29ec44b9a2bc43973c1cd14c : SUCCESS in 7m 56s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbbd81e9bd354c5a9f042d13476fb671 : SUCCESS in 6m 58s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/c9dd5761b3cd4baebbc4018ad2b24f53 : SUCCESS in 9m 08s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/10322ca1bfe0476c87cf2e3991aa15be : SUCCESS in 7m 22s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/bc5debbdb36d40418ee1ad625386b6ae : SUCCESS in 9m 01s","accounts_in_message":[],"_revision_number":17},{"id":"aa42d68209349df76291334cf98faac3c887342d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 09:47:21.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"1e585e3a669aaf63b8dbc70b3d7542e42ac30a8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 10:01:35.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fc3a6333e7bc4f96980c9ee173a31052 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59df1e9e8534469aa5b86ffabf012580 : SUCCESS in 5m 22s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/771d689b78734dedafa56759eb31a4f9 : SUCCESS in 5m 34s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/40f8a89ddccf452d9674c9a3abf7edf5 : SUCCESS in 5m 00s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/105bfcf9d5c2414089cc1efdc0a275f9 : SUCCESS in 4m 29s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/3ce57dcfb46a4dafb292b33652afe9d9 : SUCCESS in 4m 44s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/2f9d9e5749a6440490647b2c5a3c9846 : SUCCESS in 4m 38s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/931c3c9c577848d8a97a683fa72b6c8d : SUCCESS in 4m 24s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/00bba51c1f7b4899b8b2090e27748ffa : SUCCESS in 7m 23s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/7ca21b5e57ec40018894bfbea51b597e : SUCCESS in 7m 45s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3dd0dd2f2dc40309dad16b1e0f8d329 : SUCCESS in 7m 37s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dae7739386904f0d8b018fbe5af1b0d6 : SUCCESS in 6m 51s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/2eef8f5bc3444f2a873a50b740348af9 : SUCCESS in 9m 47s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/f54b0d6e0c04428285721897fc8e64c8 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":17},{"id":"1f2e7ce93a9b8147e1b828e9740f361a12e5b931","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 10:01:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"8481235950263ed391a94547bda5c9939140d031","revisions":{"99c9a4016f76d6d3f7911e2411aafea84e7e1ac9":{"kind":"REWORK","_number":1,"created":"2021-06-15 12:21:53.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/1"}}},"commit":{"parents":[{"commit":"8a1c326137e4a3cb47f84ce0255098cec31e4943","subject":"Keep using bionic for tox jobs except with Python 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8a1c326137e4a3cb47f84ce0255098cec31e4943"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"subject":"Update requirements","message":"Update requirements\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/99c9a4016f76d6d3f7911e2411aafea84e7e1ac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/99c9a4016f76d6d3f7911e2411aafea84e7e1ac9"}]},"branch":"refs/heads/master"},"4e7050c77c5a54c87bd15315a903189a0f8c2f0d":{"kind":"REWORK","_number":2,"created":"2021-06-15 12:51:12.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/2"}}},"commit":{"parents":[{"commit":"8a1c326137e4a3cb47f84ce0255098cec31e4943","subject":"Keep using bionic for tox jobs except with Python 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8a1c326137e4a3cb47f84ce0255098cec31e4943"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:50:59.000000000","tz":180},"subject":"Update requirements","message":"Update requirements\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4e7050c77c5a54c87bd15315a903189a0f8c2f0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4e7050c77c5a54c87bd15315a903189a0f8c2f0d"}]},"branch":"refs/heads/master"},"ba81f06a6df1d271ae1db11ac7711a3afeef5fec":{"kind":"REWORK","_number":3,"created":"2021-06-15 13:50:38.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/3"}}},"commit":{"parents":[{"commit":"8a1c326137e4a3cb47f84ce0255098cec31e4943","subject":"Keep using bionic for tox jobs except with Python 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8a1c326137e4a3cb47f84ce0255098cec31e4943"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 13:50:18.000000000","tz":180},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ba81f06a6df1d271ae1db11ac7711a3afeef5fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ba81f06a6df1d271ae1db11ac7711a3afeef5fec"}]},"branch":"refs/heads/master"},"b4606d9e74204d612ee98bb2e8cd33806b95663b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-15 13:50:48.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/4"}}},"commit":{"parents":[{"commit":"5c6cd7624afd46bca60761e46ceb25152e187f07","subject":"Merge \"[community goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5c6cd7624afd46bca60761e46ceb25152e187f07"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 13:50:48.000000000","tz":0},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b4606d9e74204d612ee98bb2e8cd33806b95663b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b4606d9e74204d612ee98bb2e8cd33806b95663b"}]},"branch":"refs/heads/master","description":"Rebase"},"8b242d0e7738c0f27f45a59728eb34d8207fe897":{"kind":"REWORK","_number":5,"created":"2021-06-15 13:54:37.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/5"}}},"commit":{"parents":[{"commit":"5c6cd7624afd46bca60761e46ceb25152e187f07","subject":"Merge \"[community goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5c6cd7624afd46bca60761e46ceb25152e187f07"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 13:54:31.000000000","tz":180},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8b242d0e7738c0f27f45a59728eb34d8207fe897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8b242d0e7738c0f27f45a59728eb34d8207fe897"}]},"branch":"refs/heads/master"},"8e7b364655111695f0d9c29efb1d883f71f57e1e":{"kind":"REWORK","_number":6,"created":"2021-06-15 15:02:24.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/6"}}},"commit":{"parents":[{"commit":"5c6cd7624afd46bca60761e46ceb25152e187f07","subject":"Merge \"[community goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5c6cd7624afd46bca60761e46ceb25152e187f07"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 15:01:51.000000000","tz":0},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8e7b364655111695f0d9c29efb1d883f71f57e1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8e7b364655111695f0d9c29efb1d883f71f57e1e"}]},"branch":"refs/heads/master"},"7292b25080ad80b2d4efd67f800b1f6635113415":{"kind":"REWORK","_number":7,"created":"2021-06-15 15:41:00.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/7"}}},"commit":{"parents":[{"commit":"5c6cd7624afd46bca60761e46ceb25152e187f07","subject":"Merge \"[community goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5c6cd7624afd46bca60761e46ceb25152e187f07"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 15:40:50.000000000","tz":180},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7292b25080ad80b2d4efd67f800b1f6635113415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7292b25080ad80b2d4efd67f800b1f6635113415"}]},"branch":"refs/heads/master"},"72c714f4313c9d81e42af4873f6d2b0f7a507dff":{"kind":"REWORK","_number":8,"created":"2021-06-15 15:55:49.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/8"}}},"commit":{"parents":[{"commit":"5c6cd7624afd46bca60761e46ceb25152e187f07","subject":"Merge \"[community goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5c6cd7624afd46bca60761e46ceb25152e187f07"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 15:55:39.000000000","tz":180},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/72c714f4313c9d81e42af4873f6d2b0f7a507dff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/72c714f4313c9d81e42af4873f6d2b0f7a507dff"}]},"branch":"refs/heads/master"},"81276c0f9f7abd82ed9a00b6c4bf9e9f2a80718a":{"kind":"REWORK","_number":9,"created":"2021-06-15 16:12:15.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/9"}}},"commit":{"parents":[{"commit":"5c6cd7624afd46bca60761e46ceb25152e187f07","subject":"Merge \"[community goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5c6cd7624afd46bca60761e46ceb25152e187f07"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 16:12:07.000000000","tz":180},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/81276c0f9f7abd82ed9a00b6c4bf9e9f2a80718a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/81276c0f9f7abd82ed9a00b6c4bf9e9f2a80718a"}]},"branch":"refs/heads/master"},"95aa71a170a7347d3116240cc34fe8546177a2ea":{"kind":"REWORK","_number":10,"created":"2021-06-15 16:27:42.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/10"}}},"commit":{"parents":[{"commit":"5c6cd7624afd46bca60761e46ceb25152e187f07","subject":"Merge \"[community goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5c6cd7624afd46bca60761e46ceb25152e187f07"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 16:27:37.000000000","tz":180},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/95aa71a170a7347d3116240cc34fe8546177a2ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/95aa71a170a7347d3116240cc34fe8546177a2ea"}]},"branch":"refs/heads/master"},"a54fbd782f3a2702b4a37420c70592ab84af1cfa":{"kind":"REWORK","_number":11,"created":"2021-06-15 16:50:04.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/11"}}},"commit":{"parents":[{"commit":"5c6cd7624afd46bca60761e46ceb25152e187f07","subject":"Merge \"[community goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5c6cd7624afd46bca60761e46ceb25152e187f07"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 16:49:53.000000000","tz":180},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a54fbd782f3a2702b4a37420c70592ab84af1cfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a54fbd782f3a2702b4a37420c70592ab84af1cfa"}]},"branch":"refs/heads/master"},"9a683d86c4c5bc6993bf20e67bfc71a28148ff15":{"kind":"REWORK","_number":12,"created":"2021-06-15 17:08:36.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/12"}}},"commit":{"parents":[{"commit":"5c6cd7624afd46bca60761e46ceb25152e187f07","subject":"Merge \"[community goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5c6cd7624afd46bca60761e46ceb25152e187f07"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 17:08:26.000000000","tz":180},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9a683d86c4c5bc6993bf20e67bfc71a28148ff15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9a683d86c4c5bc6993bf20e67bfc71a28148ff15"}]},"branch":"refs/heads/master"},"294537371b9379771308a38184406027d562d658":{"kind":"REWORK","_number":13,"created":"2021-06-15 17:09:33.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/13"}}},"commit":{"parents":[{"commit":"5c6cd7624afd46bca60761e46ceb25152e187f07","subject":"Merge \"[community goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5c6cd7624afd46bca60761e46ceb25152e187f07"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 17:09:28.000000000","tz":180},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/294537371b9379771308a38184406027d562d658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/294537371b9379771308a38184406027d562d658"}]},"branch":"refs/heads/master"},"0a885c1c26843a7e3e253b5863b61ce8355d5365":{"kind":"REWORK","_number":14,"created":"2021-06-15 17:59:43.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/14"}}},"commit":{"parents":[{"commit":"5c6cd7624afd46bca60761e46ceb25152e187f07","subject":"Merge \"[community goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5c6cd7624afd46bca60761e46ceb25152e187f07"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 17:59:35.000000000","tz":0},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0a885c1c26843a7e3e253b5863b61ce8355d5365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0a885c1c26843a7e3e253b5863b61ce8355d5365"}]},"branch":"refs/heads/master"},"9865110531511208486b32184f108bfacc71f323":{"kind":"REWORK","_number":15,"created":"2021-06-16 08:35:29.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/15"}}},"commit":{"parents":[{"commit":"5c6cd7624afd46bca60761e46ceb25152e187f07","subject":"Merge \"[community goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5c6cd7624afd46bca60761e46ceb25152e187f07"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-16 08:35:19.000000000","tz":0},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9865110531511208486b32184f108bfacc71f323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9865110531511208486b32184f108bfacc71f323"}]},"branch":"refs/heads/master"},"d322dbcd573e6f9ba949c3f765ce44def8cca492":{"kind":"REWORK","_number":16,"created":"2021-06-16 09:01:25.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/16"}}},"commit":{"parents":[{"commit":"eb96a5d15c2fb19c6c5a185a6f52439955be4b9d","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/eb96a5d15c2fb19c6c5a185a6f52439955be4b9d"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-16 09:01:07.000000000","tz":180},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d322dbcd573e6f9ba949c3f765ce44def8cca492"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d322dbcd573e6f9ba949c3f765ce44def8cca492"}]},"branch":"refs/heads/master"},"8481235950263ed391a94547bda5c9939140d031":{"kind":"REWORK","_number":17,"created":"2021-06-16 09:24:43.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/54/796454/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/54/796454/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/54/796454/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/54/796454/17"}}},"commit":{"parents":[{"commit":"eb96a5d15c2fb19c6c5a185a6f52439955be4b9d","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/eb96a5d15c2fb19c6c5a185a6f52439955be4b9d"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-15 12:21:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-16 09:24:39.000000000","tz":0},"subject":"Update requirements","message":"Update requirements\n\n* fix discovering os distro\n\nChange-Id: I96666afc1ad9a5c64ca5b98647cb0ba185389524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8481235950263ed391a94547bda5c9939140d031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8481235950263ed391a94547bda5c9939140d031"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
