)]}'
{"id":"openstack%2Frally~796615","triplet_id":"openstack%2Frally~master~I1c8501dd9bb46284b8aaedababe06cf379367e0d","project":"openstack/rally","branch":"master","topic":"installation_guide","hashtags":[],"change_id":"I1c8501dd9bb46284b8aaedababe06cf379367e0d","subject":"Update installation guide","status":"MERGED","created":"2021-06-16 09:41:42.000000000","updated":"2021-06-16 10:40:13.000000000","submitted":"2021-06-16 10:32:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796615-installation_guide","meta_rev_id":"ba31ff8ce66db097c36d0499eed9c2a0e0bafff2","_number":796615,"virtual_id_number":796615,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 10:32:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15086,"name":"WangXiaoFei","email":"xiaofeix.wang@icloud.com","username":"wangxf"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2021-06-16 09:45:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-16 10:40:13.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":15086,"name":"WangXiaoFei","email":"xiaofeix.wang@icloud.com","username":"wangxf"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2021-06-16 09:45:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15086,"name":"WangXiaoFei","email":"xiaofeix.wang@icloud.com","username":"wangxf"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":15086,"name":"WangXiaoFei","email":"xiaofeix.wang@icloud.com","username":"wangxf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-16 10:15:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-16 10:40:13.000000000","updated_by":{"_account_id":15086,"name":"WangXiaoFei","email":"xiaofeix.wang@icloud.com","username":"wangxf"},"reviewer":{"_account_id":15086,"name":"WangXiaoFei","email":"xiaofeix.wang@icloud.com","username":"wangxf"},"state":"REVIEWER"}],"messages":[{"id":"74541a37d773f586a9d6d45aab8a4f5a471e390f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 09:41:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e3a913992f0abdaf84a52d9a44a7fd0927bd0fc","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 09:41:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"44e1edfcc31223f0a0a5d31e9cfe58028eb75711","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 09:44:46.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1b9552a311be5b560bf210008e729cb6e4694b18","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 09:45:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0efc663f814ae69ad9ebd5eb2c361d3e72869f47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 10:15:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc3cd9179e45496fa19b14a91f20efb0 : SUCCESS in 5m 42s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/a015aa47c4a643a9812d0440be0dbb97 : SUCCESS in 5m 49s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/879b798ff5e04072a93875eb5fd9f0d5 : SUCCESS in 6m 17s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/7141765067064ee48f5201f1b96fe915 : SUCCESS in 4m 30s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/b915696842fc4dce8edbdcc090840a05 : SUCCESS in 4m 30s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/7e35bbdfaed24e39b09e10cc279f8167 : SUCCESS in 5m 45s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/351059ab0bf9418085f25c3d463352ac : SUCCESS in 4m 39s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/3d179093b18e40c19c4158b6cc22864d : SUCCESS in 4m 45s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ed33233c3bb843d685cd671281023521 : SUCCESS in 6m 56s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4314e9590db44fff8fad104bec3433e1 : SUCCESS in 6m 55s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/e8f41f6bea3b41ed81ad2b0902b0b295 : SUCCESS in 8m 45s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ba730986986426880d544f228fe57f9 : SUCCESS in 6m 48s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1ccdd333e64458c93e8cd757062d093 : SUCCESS in 6m 53s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/a8ec5bf3903d49029b050784eec65dbb : SUCCESS in 7m 29s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/e790a6f9e1874a65817ac68be5b0b043 : SUCCESS in 7m 38s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/850c806d65034b4b8c24cc2d7d7671c2 : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":2},{"id":"db3097c2f24486ec09804a76c776c1d990ab3514","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 10:15:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1c82addc7bf0499bfcc1d842e1341e70359d5b5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 10:32:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9280939aaafb4c6b93a7b60b1d3c4e1e : SUCCESS in 6m 04s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/72a31de2288d41de9d7f986cc376f9a3 : SUCCESS in 5m 36s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/c964bce82d3342a08f8247655c30c85d : SUCCESS in 5m 49s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/686c1e1ab5cd41579baea0071bb2bd34 : SUCCESS in 4m 31s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/84821e402e8c49a8983832db512c82b5 : SUCCESS in 5m 00s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/828fd654e3c0450086ec7e0e7cc8d42d : SUCCESS in 5m 17s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/383ff1da350443469b1b37731e4f7a72 : SUCCESS in 5m 03s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/2df0b3ab328946d18d94df0a7e397305 : SUCCESS in 7m 38s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/d2f5f7cc5dd943109a2909e237c95d68 : SUCCESS in 8m 51s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98289c2333d54431bd98c0c8d2e732f5 : SUCCESS in 7m 29s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62de5c65980541b5b83035a99185c1ea : SUCCESS in 7m 12s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/b72a4953b6bc42d794b93c2d2b540e92 : SUCCESS in 9m 02s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/8cef35b613834884993acd0a54e3b7b8 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":2},{"id":"9030c7a24e097a3639232405a7d02d82794e7947","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 10:32:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ba31ff8ce66db097c36d0499eed9c2a0e0bafff2","author":{"_account_id":15086,"name":"WangXiaoFei","email":"xiaofeix.wang@icloud.com","username":"wangxf"},"date":"2021-06-16 10:40:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9279809311050c832ce18825b07d745875aff31a","revisions":{"81f528f557760edfec1e5242be9c747062f08192":{"kind":"REWORK","_number":1,"created":"2021-06-16 09:41:42.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/15/796615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/15/796615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/15/796615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/15/796615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/15/796615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/15/796615/1"}}},"commit":{"parents":[{"commit":"eb96a5d15c2fb19c6c5a185a6f52439955be4b9d","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/eb96a5d15c2fb19c6c5a185a6f52439955be4b9d"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-16 09:41:34.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-16 09:41:34.000000000","tz":180},"subject":"Update installation guide","message":"Update installation guide\n\nChange-Id: I1c8501dd9bb46284b8aaedababe06cf379367e0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/81f528f557760edfec1e5242be9c747062f08192"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/81f528f557760edfec1e5242be9c747062f08192"}]},"branch":"refs/heads/master"},"9279809311050c832ce18825b07d745875aff31a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-16 09:44:46.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/15/796615/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/15/796615/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/15/796615/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/15/796615/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/15/796615/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/15/796615/2"}}},"commit":{"parents":[{"commit":"eb96a5d15c2fb19c6c5a185a6f52439955be4b9d","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/eb96a5d15c2fb19c6c5a185a6f52439955be4b9d"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-16 09:41:34.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-16 09:44:41.000000000","tz":0},"subject":"Update installation guide","message":"Update installation guide\n\nPartial-Bug: #1902703\nChange-Id: I1c8501dd9bb46284b8aaedababe06cf379367e0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9279809311050c832ce18825b07d745875aff31a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9279809311050c832ce18825b07d745875aff31a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
