)]}'
{"id":"openstack%2Frally~796695","triplet_id":"openstack%2Frally~master~I3a3c4e215724b955f168ce4f97a215aa148ff5d0","project":"openstack/rally","branch":"master","topic":"fix_contrib_link","hashtags":[],"change_id":"I3a3c4e215724b955f168ce4f97a215aa148ff5d0","subject":"Fix contribute page","status":"MERGED","created":"2021-06-16 15:19:42.000000000","updated":"2021-06-16 16:06:54.000000000","submitted":"2021-06-16 16:06:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796695-fix_contrib_link","meta_rev_id":"183db0f835d49ccf4235d076a804f6815ebb61df","_number":796695,"virtual_id_number":796695,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 16:06:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2021-06-16 15:49:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2021-06-16 15:49:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-16 15:46:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4773f470b07502f3cbc3148265328216b3e1999e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 15:19:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e356939ab42096ea125bca8edcb3aba6010dcce7","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 15:20:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"00b11384735420947c67b95bbe3c02a3233055ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 15:31:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4a93548ae9b2da17bedb2e3bacfac68416d2d4b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 15:46:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99e2881bccea4625ab3556a31aa67a41 : SUCCESS in 6m 49s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/dfad1fdedf9d40f28dca41c41f175d5a : SUCCESS in 5m 40s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/01e3e47012fa48ed8485b853d87a65f5 : SUCCESS in 5m 23s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f86c8043a464798a81235d8629491e2 : SUCCESS in 4m 55s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/88f87f5c452f4ff9b09a27733bbb4ca8 : SUCCESS in 4m 21s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/5fe38d7bb18b49b7b387089236fb07b7 : SUCCESS in 4m 22s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/0c1f2f98647c401e8a43bf348a1d1790 : SUCCESS in 4m 17s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/deee3d6b750e47e685c7ea8438328553 : SUCCESS in 4m 10s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/60cfcfac5dd248f9b296fdb3ad5f1a60 : SUCCESS in 7m 38s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/04dd81b991b2460aa386bd2fafb607b7 : SUCCESS in 7m 10s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/635dc3614f334f379268bb150acd058a : SUCCESS in 9m 08s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4323f85660d04b169c07cfd138faf9ad : SUCCESS in 7m 22s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d3a144ff246411e80932459bdfbf770 : SUCCESS in 7m 18s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/6096c24607274c13ad573c8b6d6f40f2 : SUCCESS in 8m 54s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/fdb21012bc714c8db4fc12a8b21596d8 : SUCCESS in 7m 48s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/46ef970716af482b9e2180659a0fb0c7 : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":2},{"id":"0a65014b9f14802abf68e67e13502ec13b2c773a","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-06-16 15:49:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"61966c2c3be4d7286eedd2f68d22e11451ebbc37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 15:52:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2b6a4e775ae11401cb022334f226aedfbce56dd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 16:06:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ceab61598e54b8b9d9a80eebab12b6d : SUCCESS in 6m 10s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/c12ef015abaa43849dfc1ff8cbe1cc29 : SUCCESS in 5m 30s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/3e416220a8ec4816bd6ab5cf95fc039e : SUCCESS in 5m 20s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/66882f4bdf2448d38a3a8dbf05021ae3 : SUCCESS in 4m 34s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/ae98e7f1c06f44f3b4ce085b733226ba : SUCCESS in 4m 27s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/07610bdb448042089690282bf75af8dd : SUCCESS in 5m 19s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/8b2ab67c88304d6da0a890365966a994 : SUCCESS in 5m 23s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/9f0818f2d5324d4b8c6b3c48e9052acb : SUCCESS in 7m 15s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/1efb35a86bae487e8149bb40b4428fa7 : SUCCESS in 8m 17s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1c15dc1dff54ab4a4905d47c634e754 : SUCCESS in 7m 27s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78e240ff3360463786d9cb0ccd44fa0a : SUCCESS in 7m 14s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/c0a16efa81ee409d928d7aae9ef91469 : SUCCESS in 9m 02s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/1d5a8a818fa44e7e805186d15765b2c9 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":2},{"id":"183db0f835d49ccf4235d076a804f6815ebb61df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 16:06:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1d9c1e82f33686c80e49edd041fcd8d0253afdb1","revisions":{"43f3ada2ecb4eec8cb644c3949bb5df903509568":{"kind":"REWORK","_number":1,"created":"2021-06-16 15:19:42.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/95/796695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/95/796695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/95/796695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/95/796695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/95/796695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/95/796695/1"}}},"commit":{"parents":[{"commit":"e67de57707e7f2a0300ef37c17767f8231b76aab","subject":"Prepare for a new release 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e67de57707e7f2a0300ef37c17767f8231b76aab"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-16 15:19:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-16 15:19:25.000000000","tz":180},"subject":"Fix contribute page","message":"Fix contribute page\n\nChange-Id: I3a3c4e215724b955f168ce4f97a215aa148ff5d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/43f3ada2ecb4eec8cb644c3949bb5df903509568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/43f3ada2ecb4eec8cb644c3949bb5df903509568"}]},"branch":"refs/heads/master"},"1d9c1e82f33686c80e49edd041fcd8d0253afdb1":{"kind":"REWORK","_number":2,"created":"2021-06-16 15:31:57.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/95/796695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/95/796695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/95/796695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/95/796695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/95/796695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/95/796695/2"}}},"commit":{"parents":[{"commit":"e67de57707e7f2a0300ef37c17767f8231b76aab","subject":"Prepare for a new release 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e67de57707e7f2a0300ef37c17767f8231b76aab"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-16 15:19:25.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2021-06-16 15:31:48.000000000","tz":180},"subject":"Fix contribute page","message":"Fix contribute page\n\nChange-Id: I3a3c4e215724b955f168ce4f97a215aa148ff5d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1d9c1e82f33686c80e49edd041fcd8d0253afdb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1d9c1e82f33686c80e49edd041fcd8d0253afdb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
