)]}'
{"id":"openstack%2Frally~800928","triplet_id":"openstack%2Frally~master~Id843607e7a416e15f24b820979564193f8e9ab2b","project":"openstack/rally","branch":"master","topic":"getargspec","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-05-19 18:41:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id843607e7a416e15f24b820979564193f8e9ab2b","subject":"Replace deprecated inspect.getargspec","status":"MERGED","created":"2021-07-15 11:55:49.000000000","updated":"2023-05-19 18:41:57.000000000","submitted":"2023-05-19 18:41:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800928-getargspec","meta_rev_id":"77b50dacf9b8f91dc79639803fe00fd09c217cdb","_number":800928,"virtual_id_number":800928,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-19 18:41:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2023-05-19 18:30:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2023-05-19 18:30:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-15 12:08:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-13 07:09:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-05-19 18:30:00.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"9efaf042772b127617e883a261dc9ce1665f47b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-15 11:55:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1e0b23e5484d3ffd5e22d975678877c9876778d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 12:08:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f2b369c8ee34f1d957abea62f299479 : SUCCESS in 6m 50s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/f13405d0c1264395b81713abb7cecccd : SUCCESS in 5m 55s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/234bacc9ec8748ff8ddb18a7c4536e22 : SUCCESS in 7m 18s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6efac000b51404c93ba1f25360c6d06 : SUCCESS in 4m 45s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/4d32518206e44d19bc55afe34ee5ab97 : SUCCESS in 4m 44s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/4204ca5e276f44ab8369e16f318d313e : SUCCESS in 4m 56s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/3b979d0a6d6743cba9b335fc2d6dfe70 : SUCCESS in 4m 09s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/69521f193de84e1fbe2687e10c687f0f : SUCCESS in 5m 00s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/65bac801acd148d69e36e0b622a15106 : SUCCESS in 7m 37s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f51b599be7634c9f91e2c38ac76f8d5d : SUCCESS in 7m 19s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/90fe5726e3a3415e9f88f3c4fbdb062d : SUCCESS in 7m 52s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba9f9fc19d4b4a3ebf1d99dc0f82b99b : SUCCESS in 7m 31s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f07b602b0d044e2b639eca990033dd8 : SUCCESS in 6m 46s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/6d8be0022bcd4924af4482f34697a1ba : SUCCESS in 9m 04s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/92ae8e2265d64766ad9ccf6e8b867488 : SUCCESS in 7m 55s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/f6a3dd0d352d4c3dafe2012a5cb4fac0 : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7e2f2f818f4ee5a872715e14e6a22e9026e1d5d7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-13 07:09:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26170b5f3e40ca01c11ef5bf2fa422248d17d419","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 07:20:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e5f8416c1be41c9b3db2046f9b77da6 : SUCCESS in 5m 35s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/2e89fb50bd67481b9b4341aa9a31dfa1 : SUCCESS in 5m 25s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/051adb8419574ec18da304e4e2ef5c14 : SUCCESS in 5m 36s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/88681fc1d4544aaf9ac4740b859e4d4d : SUCCESS in 4m 25s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/b17ada784f284d35b533340f0878abbe : SUCCESS in 4m 35s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/5303d4d04a6245df8a6af37c974b0b13 : SUCCESS in 5m 05s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/b2bbddf4f4344fd6b713c0c695963911 : SUCCESS in 4m 27s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/65852ce82e5c4af0a0ac0709f76ca7bb : SUCCESS in 4m 11s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/b235f556d9ba474e92de6705d8b99990 : SUCCESS in 9m 26s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a81dd9beca8a4deca83fa51d475de363 : SUCCESS in 6m 56s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/746d0f61d9f54c39a25bcbd91ac1d4a9 : SUCCESS in 8m 32s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ece366af2d284eb5b45fc755e6f746bb : SUCCESS in 7m 06s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b39b718dcd24b80bf0c44b7e2288dff : SUCCESS in 7m 31s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/7d7e651879cf4177a094599b22c7e463 : SUCCESS in 8m 32s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/60ba7c92424143b5852759d3d82aeaef : SUCCESS in 8m 48s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/22f522857828411f811d694c03019afa : SUCCESS in 8m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5f9491a0ca37dd55190bd4612326979d3f649d7b","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 18:30:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b8194327d9415b3ac516e82cb79a5ba6199b1038","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 18:30:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e3b998322f045e0354e7350d4736a0c16468a5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 18:41:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1afa5d0c29f047e681aef519df6eb47a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78a3320731814e25bdc48077a164187f : SUCCESS in 5m 04s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/d748f7532a564257b146fde4a0c68de0 : SUCCESS in 3m 39s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/28bc4f1324f44ce2acff24a699fce157 : SUCCESS in 3m 41s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/12762cc22d614434b1928d2f42887855 : SUCCESS in 2m 54s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/303401c552bf47998ba0ab460c12b066 : SUCCESS in 2m 58s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/3c85fa84576a46068b5532acf2a7c4ff : SUCCESS in 3m 56s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/04a46f4d44c44b7a987edb56b6a71a27 : SUCCESS in 4m 39s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/a5a1dc92cfc44e27955989677abf061c : SUCCESS in 3m 17s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/dea921ad8fe348399d3f5cf3a1f9ff58 : SUCCESS in 7m 22s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/0643a2ca77e04ed0a6d693042e32b7d6 : SUCCESS in 9m 00s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b67886b0e86a4692b07753035f180d81 : SUCCESS in 7m 09s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc3599f371114aaf8cb7f5516558b595 : SUCCESS in 5m 44s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/5997978a5e6d4303827bc2bf5d646fb1 : SUCCESS in 6m 52s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/2d72dbfa946c42229bcfcab5a8267157 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"77b50dacf9b8f91dc79639803fe00fd09c217cdb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 18:41:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab23f27e855be0625299e680fd2c66105eed0ffa","revisions":{"ab23f27e855be0625299e680fd2c66105eed0ffa":{"kind":"REWORK","_number":1,"created":"2021-07-15 11:55:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/800928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/28/800928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/800928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/800928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/800928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/28/800928/1"}}},"commit":{"parents":[{"commit":"1d9c1e82f33686c80e49edd041fcd8d0253afdb1","subject":"Fix contribute page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1d9c1e82f33686c80e49edd041fcd8d0253afdb1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-15 11:55:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-15 11:55:46.000000000","tz":540},"subject":"Replace deprecated inspect.getargspec","message":"Replace deprecated inspect.getargspec\n\ninspect.getargspec was deprecated since Python 3.0 and\ninspect.getfullargspec is its replacement with correct handling of\nfunction annotations and keyword-only parameters[1].\n\n[1] https://docs.python.org/3/library/inspect.html#inspect.getargspec\n\nChange-Id: Id843607e7a416e15f24b820979564193f8e9ab2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ab23f27e855be0625299e680fd2c66105eed0ffa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ab23f27e855be0625299e680fd2c66105eed0ffa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
