)]}'
{"id":"openstack%2Frally~816924","triplet_id":"openstack%2Frally~master~I06c0c4bc4f23e8255c539ff54d258ae1c7eb717b","project":"openstack/rally","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-11-15 19:38:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I06c0c4bc4f23e8255c539ff54d258ae1c7eb717b","subject":"Add rally-tox-py39 job","status":"MERGED","created":"2021-11-07 15:55:26.000000000","updated":"2021-11-15 19:38:50.000000000","submitted":"2021-11-15 19:38:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816924","meta_rev_id":"fa9e99ec32e91fe2726eb381f78fcb76bc45e4a1","_number":816924,"virtual_id_number":816924,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-15 19:38:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2021-11-15 19:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2021-11-15 19:23:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-07 16:15:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-15 19:23:37.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"12ff4df1e7e2003ba7657be69dfa685da09fe3d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-07 15:55:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbe45e24eb52ed783431bcbf70662075d8d773ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-07 16:15:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8986c2341d0d45ec983bedc9c3841885 : SUCCESS in 5m 46s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/3502f26fb2e640ac928364b8803ed826 : SUCCESS in 5m 31s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/e2c61045c4bf4939a7dcdc52aac6fc20 : SUCCESS in 6m 03s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/87bb8b01bd0c49e893afe028a5380752 : SUCCESS in 4m 24s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/f8a123ca95554b7cb33c69c55afa540c : SUCCESS in 4m 05s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/a26284c18fdd49a2915684a8cefa0cac : SUCCESS in 4m 41s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/1bfd18b6136a4f3bb9f7d7d151a8a8c7 : SUCCESS in 3m 51s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/6f40a27505ef4b008b9ddfbe8b88a179 : RETRY_LIMIT in 2m 53s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/8a1a02e63d3f4f70bc0c830b5e974d30 : SUCCESS in 4m 24s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/785cf3b0d12849fc9d9c59ffd380ee3f : SUCCESS in 7m 25s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dd0a9a00c28d4a70a85472d0010c4f55 : SUCCESS in 7m 21s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/c12e900a329b44759e7167bc1c456f67 : SUCCESS in 6m 59s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10207a91579e4b54822d4531fa9a3cb2 : SUCCESS in 7m 04s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba7a76c016c443e9b3badc6da21b4c8e : SUCCESS in 7m 36s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/c663994dbcbb4328897515abd64939c3 : SUCCESS in 8m 45s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/28e4de2b60b84846bd18e2a747f97b3d : SUCCESS in 8m 23s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/af7eecfabe1b46139983aec258e7cdcd : SUCCESS in 7m 56s","accounts_in_message":[],"_revision_number":1},{"id":"38450d493bd0d99be2b73dad6395cb07256ada75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-11 15:19:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b10bb5f98848f3124f2485660477c14e0d639a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 16:39:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa5bd7ae918426aa7845d47c13c6d4f : SUCCESS in 6m 46s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/0c8a61cb3eb042d489dfc02f1ae2831c : SUCCESS in 8m 18s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/dd621a148d0d4221bd2bf235572995f4 : SUCCESS in 5m 58s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4d50c620c1a4358b0f135b08667b7a9 : SUCCESS in 6m 18s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/810bc37e66fa44e5a0247fcfdc8a444c : SUCCESS in 7m 53s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/27412d4b58f345a4b09a9733c3fdd4c7 : SUCCESS in 6m 42s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/44ec929cfdf044c49e4e6f161ee84c78 : SUCCESS in 6m 28s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/f41addcec91544c0abe2d50de0724253 : SUCCESS in 6m 29s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/ee35d09771334dccb2288781f78270d1 : SUCCESS in 5m 46s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/bbb3b9e4792841f28b78dcc9eccec306 : SUCCESS in 9m 47s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/40ddc1d981d443638531aaa1d2c2582f : SUCCESS in 11m 16s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/3ed9e6a7091340fb9a49e8f0764c0262 : SUCCESS in 9m 12s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbc1d6b171dd4cef8801be0b2d57fa00 : SUCCESS in 8m 15s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3f918e2a9a94ed1a8f43e4a004c844c : SUCCESS in 9m 08s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/9f57310cd35b486c8d56b6c2def086cb : SUCCESS in 10m 40s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/bec2bee9da62437385d47867c7ce4205 : SUCCESS in 8m 55s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/a34a05259b5848d196eeabef29600a3e : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":2},{"id":"e8cc659acec9e2a18da860b33e57ebc4ac1e2df9","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2021-11-15 19:23:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"241fa117a576a0fb3af632445208c1696887c48a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 19:24:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"18465c550eef652f9c2f0e8f68a22bf8ad964175","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 19:38:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5cad12b8b934ec79dcf56a21223c472 : SUCCESS in 6m 12s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b6382b89a77b41c9bd2dfc3d5385d590 : SUCCESS in 5m 27s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/dd0d04e743934b7e85c130b44520787c : SUCCESS in 5m 26s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aa8f1a82db240d58d5cf31b70fa8b9b : SUCCESS in 4m 47s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/adea7d29ad2f40888236960e574407c7 : SUCCESS in 4m 53s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/8d16aeee7972498cbfbc7822209b6139 : SUCCESS in 4m 54s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/46f6a05a035a4437b327273ee625409d : SUCCESS in 5m 17s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/07772fb9af154ef9885b67183a55769b : SUCCESS in 5m 04s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/0a8e29b933004dbda4e1977c24b0a1a3 : SUCCESS in 8m 56s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/816493349d6544048d0243685b14cc0b : SUCCESS in 8m 24s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b13813832f043b5b12410c024e1f776 : SUCCESS in 7m 49s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59dcb9196be64c7382d21523f3a5b148 : SUCCESS in 7m 31s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/5a5467aae0fb42449787d3eaa07c9ce2 : SUCCESS in 8m 34s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/1a77901321a64ba7821e8dc9f9fccb83 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":2},{"id":"fa9e99ec32e91fe2726eb381f78fcb76bc45e4a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 19:38:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0f0f20dab2cbb3f519077df9f53600e81f93ad42","revisions":{"4a7977b143a52d67eceb80a9c1274ae89983589e":{"kind":"REWORK","_number":1,"created":"2021-11-07 15:55:26.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/24/816924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/24/816924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/816924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/816924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/816924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/24/816924/1"}}},"commit":{"parents":[{"commit":"3c43987344836f8668bbbeea9e9f71df79b96ca9","subject":"Moving IRC network reference to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3c43987344836f8668bbbeea9e9f71df79b96ca9"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-11-07 15:53:05.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-11-07 15:55:11.000000000","tz":60},"subject":"Add rally-tox-py39 job","message":"Add rally-tox-py39 job\n\nPython3.9 has been added by the TC to the list of supported python\nversions, so it will be nice to test against it.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I06c0c4bc4f23e8255c539ff54d258ae1c7eb717b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4a7977b143a52d67eceb80a9c1274ae89983589e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4a7977b143a52d67eceb80a9c1274ae89983589e"}]},"branch":"refs/heads/master"},"0f0f20dab2cbb3f519077df9f53600e81f93ad42":{"kind":"REWORK","_number":2,"created":"2021-11-11 15:19:56.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/24/816924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/24/816924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/816924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/816924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/816924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/24/816924/2"}}},"commit":{"parents":[{"commit":"3c43987344836f8668bbbeea9e9f71df79b96ca9","subject":"Moving IRC network reference to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3c43987344836f8668bbbeea9e9f71df79b96ca9"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-11-07 15:53:05.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-11-11 15:19:32.000000000","tz":60},"subject":"Add rally-tox-py39 job","message":"Add rally-tox-py39 job\n\nPython3.9 has been added by the TC to the list of supported python\nversions, so it will be nice to test against it.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I06c0c4bc4f23e8255c539ff54d258ae1c7eb717b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0f0f20dab2cbb3f519077df9f53600e81f93ad42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0f0f20dab2cbb3f519077df9f53600e81f93ad42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
