)]}'
{"id":"openstack%2Frally~823996","triplet_id":"openstack%2Frally~master~I94f597d99951459b12f0f0211ec73f2ae7fa908d","project":"openstack/rally","branch":"master","topic":"fix_open_descriptors_at_ssh","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2022-01-10 14:41:50.000000000","reason":"Andrey Kurilin replied on the change"}},"hashtags":[],"change_id":"I94f597d99951459b12f0f0211ec73f2ae7fa908d","subject":"Close stringio objects at sshutils","status":"MERGED","created":"2022-01-10 14:06:26.000000000","updated":"2022-01-10 14:58:41.000000000","submitted":"2022-01-10 14:58:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823996-fix_open_descriptors_at_ssh","meta_rev_id":"750853cb18450794ec3200aa271483be91d0450d","_number":823996,"virtual_id_number":823996,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":10267,"name":"venkata anil","email":"anilvenkata@redhat.com","username":"anil"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-10 14:58:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2022-01-10 14:41:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":1,"date":"2022-01-10 14:31:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10267,"name":"venkata anil","email":"anilvenkata@redhat.com","username":"anil"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2022-01-10 14:41:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":10267,"name":"venkata anil","email":"anilvenkata@redhat.com","username":"anil"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":10267,"name":"venkata anil","email":"anilvenkata@redhat.com","username":"anil"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-10 14:31:47.000000000","updated_by":{"_account_id":10267,"name":"venkata anil","email":"anilvenkata@redhat.com","username":"anil"},"reviewer":{"_account_id":10267,"name":"venkata anil","email":"anilvenkata@redhat.com","username":"anil"},"state":"REVIEWER"},{"updated":"2022-01-10 14:34:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd0380ea528be8a1f48d660f5cfcdb7593442d2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2022-01-10 14:06:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40e26a87cd4bde25d89d09f36d693ffe5a286861","author":{"_account_id":10267,"name":"venkata anil","email":"anilvenkata@redhat.com","username":"anil"},"date":"2022-01-10 14:31:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bdae8b8bc684fc6b210defa62401cbbcce53f6f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 14:34:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a2880e192f04b988926fbecf72cdf07 : SUCCESS in 6m 27s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/404aae26474049908d0c40fd4917362f : SUCCESS in 5m 15s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ea9355ed408445fbab50a85f88c51cc0 : SUCCESS in 5m 14s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/c271aec4547d4cde94d71adfcd252744 : SUCCESS in 4m 31s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/5b31c8fc5d024cde87359f02927de222 : SUCCESS in 5m 39s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/d131c9e197344cbfb0dca1236d6e4364 : SUCCESS in 5m 22s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/a935757ba12640108e8b08f502658ba3 : SUCCESS in 4m 46s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/9c223f8f040d4e73bbdc3ff95edbc269 : SUCCESS in 5m 03s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/010e5a46b21b4e1689ea18402478873a : SUCCESS in 5m 05s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/859e01201d9d4d518686a04bd4dddfde : SUCCESS in 8m 42s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2b7a556caef14bf5bc955cff4a7348c1 : SUCCESS in 9m 08s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/c8f82b5657e94954b80e19bf71cb721e : SUCCESS in 8m 20s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/179067edc1854d64aef64d1c27a3544f : SUCCESS in 7m 42s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34e6556fff5140aaa0b243644dad11de : SUCCESS in 7m 34s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/9d20d4c1b4b746c9819b0cdedd5fc788 : SUCCESS in 9m 45s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/8f462c2332854acf852153fc9111dba1 : SUCCESS in 9m 07s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/ee6287b65a314deca8656009e4d2a3fb : SUCCESS in 9m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f184156ca15e36c9d4a4db90110d7f3246f828e0","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2022-01-10 14:41:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e6666f4d0bcc0ee72b81eb2527477f8f2dfa5084","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 14:42:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d750056f991523c80665a8892b2f21cdddf9431f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 14:58:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e3658519c3d4523ac17bd227686631b : SUCCESS in 5m 40s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/d401c1b63ad246e5a25cf2c95a867992 : SUCCESS in 6m 09s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/00876e4643cc4653b3f1fd252a31b51f : SUCCESS in 6m 00s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/c80fb6fa32de4a379b445268544e6ce1 : SUCCESS in 6m 08s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/baf5d8ca0c814302a09e45225491d9d2 : SUCCESS in 5m 48s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/f697f5a3412a4bc1b91d8986b77a02f2 : SUCCESS in 5m 21s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/7b7eb410e23b404c8e45487dd6c999d5 : SUCCESS in 5m 00s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/8eaa9e52ee144c82964cd4dc57d9526e : SUCCESS in 5m 35s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/76cfda76c54a44768dd365fa6755fc96 : SUCCESS in 7m 44s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/d447035185354137badcceaa503da186 : SUCCESS in 8m 58s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91767abf347b4e14bc9ade664eb839e1 : SUCCESS in 7m 26s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7facae717126487f8413ffb11e8311b1 : SUCCESS in 6m 59s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/0bc546c6188b4e6aae74dcdc002d4343 : SUCCESS in 8m 52s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/5b8cbd00160b4b23ae0fabaf487ce208 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":1},{"id":"750853cb18450794ec3200aa271483be91d0450d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 14:58:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a77c63071abcf2e2088b72f1a81b3ffc69be7c7","revisions":{"2a77c63071abcf2e2088b72f1a81b3ffc69be7c7":{"kind":"REWORK","_number":1,"created":"2022-01-10 14:06:26.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/96/823996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/96/823996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/823996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/823996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/823996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/96/823996/1"}}},"commit":{"parents":[{"commit":"675ea00d3b5afc410a2f9d93de6258ef08a2f560","subject":"Update doc to do not refer to old installation script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/675ea00d3b5afc410a2f9d93de6258ef08a2f560"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-01-10 13:38:11.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-01-10 14:05:39.000000000","tz":120},"subject":"Close stringio objects at sshutils","message":"Close stringio objects at sshutils\n\nCloses-Bug: #1956956\nChange-Id: I94f597d99951459b12f0f0211ec73f2ae7fa908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2a77c63071abcf2e2088b72f1a81b3ffc69be7c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2a77c63071abcf2e2088b72f1a81b3ffc69be7c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
