)]}'
{"id":"openstack%2Frally~824172","triplet_id":"openstack%2Frally~master~Ief328c82f43d4717880da17b62cb4a089e1ed9cf","project":"openstack/rally","branch":"master","topic":"pip_install","hashtags":[],"change_id":"Ief328c82f43d4717880da17b62cb4a089e1ed9cf","subject":"[ci] force reinstall of pip","status":"MERGED","created":"2022-01-11 12:42:45.000000000","updated":"2022-01-11 14:37:06.000000000","submitted":"2022-01-11 14:37:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824172-pip_install","meta_rev_id":"88b817fbc0299015627473426a566a451a4b0ac6","_number":824172,"virtual_id_number":824172,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-11 14:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2022-01-11 14:21:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2022-01-11 14:21:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-11 12:57:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a6e94338df3df8ff9b9e101cd7ed987ba005114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2022-01-11 12:42:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb9a10a07a08c89f9280f5db907cf682eec04446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 12:57:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03755f0330dd487b9a57348f62640aeb : SUCCESS in 5m 44s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/21d0b628be884f748a69e4229eb59d71 : SUCCESS in 5m 53s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/b801c085d0e643a4be5e334bd2954e4e : SUCCESS in 5m 15s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/37d394479d4349bbbfff1e85a24917f9 : SUCCESS in 4m 41s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/52951ab971e74b28ac7b3e568956e98f : SUCCESS in 5m 26s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/1681b5c533904a67814df5b6cdd43d34 : SUCCESS in 4m 48s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/28ccfe88bb234748a6dd1eccdac5ca3e : SUCCESS in 4m 33s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/c30e1fda160e436489b69bb34dd7efa9 : SUCCESS in 6m 08s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/8e456b7e7ac842b584c398794cb28047 : SUCCESS in 4m 20s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/da54f9b046db4632ab0ce3d6547521c3 : SUCCESS in 8m 47s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/76c20f2628ce43698e39bd5bfb5931ad : SUCCESS in 6m 59s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/5355a9b1583042268cb72fcc73a16c01 : SUCCESS in 8m 04s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7b1b696a0f9489d8adda20504e9702c : SUCCESS in 7m 41s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5130970afc2a4999bb36ac05eede1163 : SUCCESS in 7m 03s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/44509537f3f84d6f8dc957c9b12b5ae4 : SUCCESS in 8m 28s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/6e52e7e943bb4469bd53d5f1fdb88f90 : SUCCESS in 7m 38s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/7e62cbbdd09b4da58d346410846c2efd : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c4be7d3fb6366c61911667382251fb7f61b48315","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2022-01-11 14:21:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca62e66a1735104a19c41ed736a94b8cc5b25b79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 14:22:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bdf363a30123be5b150c277aa8089f4bc9720f9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 14:37:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b608557de83a4f48b32652922fa04982 : SUCCESS in 5m 52s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/9b4fe6a95e824ebab522ff6872e19b75 : SUCCESS in 5m 58s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/cfdff246e0214a69aa34b4634e43b60a : SUCCESS in 5m 42s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/a58544526b604acb8368e2271a632e37 : SUCCESS in 7m 35s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/d0dcfcbe784f49308c7906a2e70bdd41 : SUCCESS in 7m 15s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/fa3ee7922a8a4036811bb28efa096c4d : SUCCESS in 5m 47s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/2c11f5584f7a4ff68eaaa3632f52f31b : SUCCESS in 5m 16s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/68a4b745ea2c4192982c422fb5814f10 : SUCCESS in 7m 00s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/3d07ac8b7057459eb04d25cc06fbe7c3 : SUCCESS in 10m 39s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/f427b95a1b8449bfb5a6139531938fc5 : SUCCESS in 8m 06s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c86c792301f41bb81264b40b7fe4bcf : SUCCESS in 9m 30s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e2a2ce0f2c8464a89d565a3f686a8fe : SUCCESS in 8m 58s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/3c9ebba92c244b7792406c5f72b07204 : SUCCESS in 11m 06s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/cd223b8843844e6c810c73c7c2f2efe5 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":1},{"id":"88b817fbc0299015627473426a566a451a4b0ac6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 14:37:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72859c8f6793ad55ee5528cd7d297ddb7e62dfcf","revisions":{"72859c8f6793ad55ee5528cd7d297ddb7e62dfcf":{"kind":"REWORK","_number":1,"created":"2022-01-11 12:42:45.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/72/824172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/72/824172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/72/824172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/72/824172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/72/824172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/72/824172/1"}}},"commit":{"parents":[{"commit":"2a77c63071abcf2e2088b72f1a81b3ffc69be7c7","subject":"Close stringio objects at sshutils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2a77c63071abcf2e2088b72f1a81b3ffc69be7c7"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-01-11 12:41:55.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-01-11 12:42:03.000000000","tz":120},"subject":"[ci] force reinstall of pip","message":"[ci] force reinstall of pip\n\nChange-Id: Ief328c82f43d4717880da17b62cb4a089e1ed9cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/72859c8f6793ad55ee5528cd7d297ddb7e62dfcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/72859c8f6793ad55ee5528cd7d297ddb7e62dfcf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
