)]}'
{"id":"openstack%2Frally~824362","triplet_id":"openstack%2Frally~master~I6d181c6aa3568c05d2af9bf9f7bf975d7be32f0b","project":"openstack/rally","branch":"master","topic":"call-pip-different","hashtags":[],"change_id":"I6d181c6aa3568c05d2af9bf9f7bf975d7be32f0b","subject":"Call pip install tox differently","status":"MERGED","created":"2022-01-12 09:25:49.000000000","updated":"2022-01-12 10:30:55.000000000","submitted":"2022-01-12 10:30:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824362-call-pip-different","meta_rev_id":"8a5e62037aa9b53f6c126497e5bc11b88ce1dfa5","_number":824362,"virtual_id_number":824362,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-12 10:30:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2022-01-12 10:14:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2022-01-12 10:14:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 09:48:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0c9703e5be6e5ddb69770428392fc5837afdc82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2022-01-12 09:25:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad1b9adb51d41df1edafd303e9c022a1bef489a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 09:48:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea893b914dc3462d8b01eb44852e0b61 : SUCCESS in 8m 25s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e6d168ea1093403f8e3abd45014e6671 : SUCCESS in 8m 16s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/c83b3892c58b4a4c8658f2c550529868 : SUCCESS in 5m 08s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdef9d7f63834d819eb64a3c312c8352 : SUCCESS in 7m 12s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/9d9f43dea83846cf9ebe6a51c3a34d2a : SUCCESS in 4m 48s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/a89dd1ee629f46839f929b7dcffe9fd9 : SUCCESS in 8m 32s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/bce83be613dc4afa813c977bc49e71e9 : SUCCESS in 6m 06s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/49efc9e62efc4f5bb8d7aced587c3ea7 : SUCCESS in 5m 31s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/a3746bfd3d7d4ac4a540f7d0794f1090 : SUCCESS in 5m 02s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/d66f6d2eb1aa4509be58e8cc3e42ac1d : SUCCESS in 8m 07s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9eec7559f6114debb85dde10319ddc0c : SUCCESS in 11m 01s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/febea9fa993846e58f05198a89df8b89 : SUCCESS in 8m 41s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84dc1cb57c164c16b0053cc8818a4577 : SUCCESS in 7m 35s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3579faaa1124ee7ba6f348ffa8d6fee : SUCCESS in 9m 49s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/dea94f1f939b46339169d9319d5b89e0 : SUCCESS in 10m 10s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/abf6a06cf7b3441b9fa4754a662f8124 : SUCCESS in 9m 23s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/bfaed68faec24b51b48e1a4a4493a3a9 : SUCCESS in 8m 31s","accounts_in_message":[],"_revision_number":1},{"id":"18d4e73641892e59be308fb7454badc172f37556","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2022-01-12 10:14:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7911df42f3058356fc4800bbe0d147ad7221ef85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 10:15:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02aa4b026023ba776fd74444d7c73e21ebf9d05a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 10:30:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9874fadf9b13409ea202729d35a8e1da : SUCCESS in 7m 24s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/84d6dfa662444e59b4fca87912581792 : SUCCESS in 7m 06s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/5da4e6c4a62442469918d03de18964f6 : SUCCESS in 6m 53s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/a79d9b1486d24370a2ec806cbd9fd2da : SUCCESS in 4m 41s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/164ef81a0b1d413d9329f7779963cf09 : SUCCESS in 5m 31s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/90fac46af59b4eef902855f03f246a41 : SUCCESS in 5m 24s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/ff5595e899d94ca0a51d98a1196ec2a7 : SUCCESS in 6m 13s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/6a79df0ca42b475a89ab35203d774c3f : SUCCESS in 4m 59s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/d6d2c4ac5fbb436b8028d755eaebb9c5 : SUCCESS in 8m 47s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ce7bd453abf2422bb7039e5d3e2003a3 : SUCCESS in 8m 00s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/433738d0a4704892a27355141370f9fb : SUCCESS in 7m 26s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a26ef3e5214b496b97024504cd5ddd42 : SUCCESS in 7m 10s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/411559590e1541a994a7cd173c40a78f : SUCCESS in 9m 50s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/e20211ccf06b43bf8d768780b7e93802 : SUCCESS in 9m 11s","accounts_in_message":[],"_revision_number":1},{"id":"8a5e62037aa9b53f6c126497e5bc11b88ce1dfa5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 10:30:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58b12c2e0bfa541bdb038be6e1679c5117b98484","revisions":{"58b12c2e0bfa541bdb038be6e1679c5117b98484":{"kind":"REWORK","_number":1,"created":"2022-01-12 09:25:49.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/62/824362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/62/824362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/824362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/824362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/62/824362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/62/824362/1"}}},"commit":{"parents":[{"commit":"72859c8f6793ad55ee5528cd7d297ddb7e62dfcf","subject":"[ci] force reinstall of pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/72859c8f6793ad55ee5528cd7d297ddb7e62dfcf"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-01-12 09:25:16.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-01-12 09:25:16.000000000","tz":120},"subject":"Call pip install tox differently","message":"Call pip install tox differently\n\nChange-Id: I6d181c6aa3568c05d2af9bf9f7bf975d7be32f0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/58b12c2e0bfa541bdb038be6e1679c5117b98484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/58b12c2e0bfa541bdb038be6e1679c5117b98484"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
