)]}'
{"id":"openstack%2Frally~824851","triplet_id":"openstack%2Frally~master~I6292cc2f0ce7038a5426bbb9594ba10d3fe0cd16","project":"openstack/rally","branch":"master","attention_set":{},"removed_from_attention_set":{"33881":{"account":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"last_update":"2023-12-09 09:47:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6292cc2f0ce7038a5426bbb9594ba10d3fe0cd16","subject":"Update python testing classifier","status":"ABANDONED","created":"2022-01-17 02:40:56.000000000","updated":"2023-12-09 09:47:06.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2fe335a1fc57a49f9cf1fb72f80e6e7717401501","_number":824851,"virtual_id_number":824851,"owner":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-11 13:18:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-17 03:03:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-24 15:35:51.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"CC"},{"updated":"2023-03-11 12:53:36.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"CC"}],"messages":[{"id":"80e59bb6683f217e210eff457293d9bb96ae4c50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-01-17 02:40:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcbbf4e86028e997065007227ced5e9ce7a038bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 03:03:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5e7a35221a914c4486200594249c6d57 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5863a5e23a4b46718a1e124983c87192 : SUCCESS in 7m 57s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/367da151d39f4994a0505160cdb09e6b : SUCCESS in 7m 24s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/0568beb8bcca49b5a431f549385e24ca : SUCCESS in 7m 37s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/491d179f632a42039f0f37f934da83b2 : SUCCESS in 5m 55s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/e8786759d89345a6938563f5e4fe2a29 : SUCCESS in 5m 48s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/08c6ad2d769d4ae995dcd447b71db1f2 : SUCCESS in 6m 05s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/b22a93e86bee425fb571604bc9b03e27 : SUCCESS in 5m 26s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/743420ca8f9f44d4b148d4e37e3cbded : SUCCESS in 5m 33s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/ef3d88b16f5c402f821ea665e2e9e095 : SUCCESS in 4m 25s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/e2b09328aaa24500b056f1083d195049 : SUCCESS in 8m 35s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b0172d8e060c499bb717815bd4baea7b : SUCCESS in 7m 57s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/21b23502792d41f0aa7dae42bcc9f8ec : SUCCESS in 8m 29s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72aa502d4457469e9a45eecba8e10304 : SUCCESS in 8m 17s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/197fda914d774d1b90891ec80c5d82f6 : SUCCESS in 7m 50s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/baa738d22c224400b27d5ae6ca2f7c2d : SUCCESS in 9m 40s\n- rally-install-centos-8 https://zuul.opendev.org/t/openstack/build/bbad9418c0604885843fe6e812067b38 : SUCCESS in 8m 21s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/d2c033bb715a41a680047e55b0a5e55e : SUCCESS in 9m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2a9c78961d9c93974723dffa61050ecd37b88c30","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2022-10-24 15:35:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"950d60d376df1d5603dcce4210769b385d0190dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 15:48:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c50177fd6d74200892cdf6beaa55bea\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4110efbbf8874c5b91caa65d1916712e : SUCCESS in 1m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d724d5a6cd34d83a35892143a5e8ba3 : SUCCESS in 5m 51s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/d8f35a352a8544c4987fea755ec39f27 : SUCCESS in 4m 19s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/3e5ae870df6a4f669d8fc573ed3f7b92 : SUCCESS in 4m 05s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/49fbb6a10e3d4f73b4cc2d8cd50e647a : SUCCESS in 3m 09s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/68dcff4ad8064044b1e66550889f859e : SUCCESS in 3m 38s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/12b68dc355de43e99b55bb9b99834077 : SUCCESS in 4m 09s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/b1355b8108fa4e65a500feab7bcff8eb : SUCCESS in 3m 27s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/2d4c1cbaf62242009b93c9baa5612d20 : SUCCESS in 3m 19s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/d7c5a3a1ff7b461f9bf2bde9cf154627 : SUCCESS in 3m 00s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/398046b5c4234280b5e24393d1d570de : SUCCESS in 6m 53s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3e6c025464bc4cd18c8ce8dbc3ac8df6 : FAILURE in 6m 38s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/1cfa6d0e2abe4baab771c20c60967481 : SUCCESS in 6m 22s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a49d77567f3c401cb57650fc517657c3 : SUCCESS in 6m 06s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03bf1f4b52cc49a0842481a298889a24 : SUCCESS in 5m 52s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/74e7844d0d98407abd643caf0cf43b37 : SUCCESS in 6m 59s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/98c3a4cedf8c454b953b96b052262398 : SUCCESS in 7m 32s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/b8530468d3be4c088ff1bd59e47491e8 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"b2f2aa993ace9a379be3b0b2731df6cb7c67d26d","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-03-11 12:53:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"21b6f21d908d47d70e014f8621516ce566ebcff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 13:18:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7551e57836a44e1094f1830754fac869\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9ed6268d3ffe4ec499e92e85619c8632 : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cdbfc7d41834597b88cb1daf963c5e2 : SUCCESS in 4m 22s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/09dbdaa3ac494920b2a9840cd3b686a4 : SUCCESS in 5m 57s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/a292d9f5ea4b4026b2d90aa114fa9eb2 : SUCCESS in 3m 50s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/301cf70a57684a1fbd73e04170ca03d7 : SUCCESS in 2m 51s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/43fd849345a04387bbc6883cf4327619 : SUCCESS in 2m 59s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/0583653b74ea444b8bb8e4c0fe4f1ed4 : SUCCESS in 2m 59s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/ef83bf21e0604f0f91fe0efe6520b033 : SUCCESS in 2m 50s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/143f609445a24cac8c2e64b0b6351d11 : SUCCESS in 2m 58s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/061997753d3c4862821715c9cdb7f538 : SUCCESS in 3m 06s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/3dbe0b72949f4d2c90ca995b39e395dd : SUCCESS in 5m 13s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2c1378d055ec4a02a313f0331804025e : SUCCESS in 5m 12s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/7a361c0e304d4f83b4c81a2db000f983 : SUCCESS in 6m 27s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecaee9374556459d95782cd12b465bdd : SUCCESS in 5m 48s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d550cda29dd74b2aa322b82ea206ad3a : SUCCESS in 5m 25s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/9c87bd739e384225acabb64278bbdd4f : SUCCESS in 6m 20s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/b5d4839276274355adadc56262709224 : SUCCESS in 7m 15s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/2e737046569b40899209bc8f836a3171 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2fe335a1fc57a49f9cf1fb72f80e6e7717401501","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-12-09 09:47:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cccdd7e69c0d3365d3b9a5b1d6473b18a4a190c6","revisions":{"cccdd7e69c0d3365d3b9a5b1d6473b18a4a190c6":{"kind":"REWORK","_number":1,"created":"2022-01-17 02:40:56.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/51/824851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/51/824851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/51/824851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/51/824851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/51/824851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/51/824851/1"}}},"commit":{"parents":[{"commit":"a188010eb94875eb7592f343ec2f6addb489fd89","subject":"remove unicode from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a188010eb94875eb7592f343ec2f6addb489fd89"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-01-17 02:38:47.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-01-17 02:38:47.000000000","tz":480},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nChange-Id: I6292cc2f0ce7038a5426bbb9594ba10d3fe0cd16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cccdd7e69c0d3365d3b9a5b1d6473b18a4a190c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cccdd7e69c0d3365d3b9a5b1d6473b18a4a190c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
