)]}'
{"id":"openstack%2Frally~852195","triplet_id":"openstack%2Frally~master~I5904b75b0c60d8fd99e4fb57b752109b960cff45","project":"openstack/rally","branch":"master","topic":"docutils-warning","hashtags":[],"change_id":"I5904b75b0c60d8fd99e4fb57b752109b960cff45","subject":"Suppress deprecation warning from docutils","status":"MERGED","created":"2022-08-04 16:40:38.000000000","updated":"2022-08-04 17:10:10.000000000","submitted":"2022-08-04 17:10:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852195-docutils-warning","meta_rev_id":"ea37459564257dc24213a9b985f0b5815930a010","_number":852195,"virtual_id_number":852195,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-04 17:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2022-08-04 16:40:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2022-08-04 16:40:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 16:55:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"28c89ec7792c321a6a85e802f474831e1e791edd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2022-08-04 16:40:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ad40eaf54c9745c7b4dfb40c5e781e72fff3c27","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2022-08-04 16:40:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d24875abc6c09e6c699f50f00e5ac3c8c29ff8a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 16:55:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aeb82d6f1ff4d8da1b1fa888aaa5f8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56916fd122d242c58ff77a7ccc77d437 : SUCCESS in 5m 20s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/cc06a6dcabd7497799bc0509c0c7b223 : SUCCESS in 5m 29s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/088e6ec719784e7a87f886bd9098d7c2 : SUCCESS in 5m 15s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f5926b24a2d471085a2e1e3b17a8e4c : SUCCESS in 4m 04s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/c3726030130345d2ae9f13faa5a68241 : SUCCESS in 4m 17s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/a1edec9d868b410082e07646cab2063f : SUCCESS in 3m 58s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/2478f06df68544b184cc57e4a228e945 : SUCCESS in 4m 28s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/e482c7d9a3ec45308018f96888fbf057 : SUCCESS in 3m 52s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/50f00882980b4a7fb949c0372de8b6e0 : SUCCESS in 4m 12s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/d365c7d670a54c509a22274cf70a7226 : SUCCESS in 7m 02s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1692776512b94cb38c4377c5b420ecef : SUCCESS in 6m 27s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/fe87b72d151a49ddb916d7a24a100193 : SUCCESS in 6m 42s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eff2296cddac4871a26198418c6dd9db : SUCCESS in 6m 44s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/814efdf261f54fabb99803b22efd57ff : SUCCESS in 6m 22s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/4c522dde4b9447e5b2b385029720a902 : SUCCESS in 7m 32s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/a720dca59df248849db57d948b687ae1 : SUCCESS in 7m 50s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/5d7e3667ea1a4d3489eb5da0eb8f262a : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":1},{"id":"7542e1d6633415b893b84bca112e83538985e4f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 16:55:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb0b934a3085b42351ce3498594fb17ea7644754","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 17:10:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d30dab61d0d4a7ebe8c9e358da8be46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a44ffa0ac8ae4cb28f0541ad5bcb1e86 : SUCCESS in 5m 16s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/11f701848dc6420dabc94162d57985fb : SUCCESS in 4m 57s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/82771eb5da424ab1a51e647462e15177 : SUCCESS in 4m 29s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c28d82033cc43a6b8931c705c7b6d95 : SUCCESS in 3m 32s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/fcfb609ed01b41ffbedaf6c06ee88bbd : SUCCESS in 3m 40s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/16028e15cd094f8db728ec37d6d9f2e3 : SUCCESS in 3m 51s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/d7f4126e46e54786bff98f4932e3658a : SUCCESS in 3m 41s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/3141cf0d848d45e091506bea3435662f : SUCCESS in 3m 42s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/a424a972663a468fb58b47a589ebff5b : SUCCESS in 5m 59s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/3f3279d35ad74e26afa100047551a547 : SUCCESS in 6m 48s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22d1da47d47c49919275b028b5932b77 : SUCCESS in 6m 26s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e56d669864e44245810801bd4013d526 : SUCCESS in 6m 15s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/4a6ad0ff7a644a319247b02474257a65 : SUCCESS in 7m 43s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/51875ff788af480bbf7160b131ab259d : SUCCESS in 8m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ea37459564257dc24213a9b985f0b5815930a010","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 17:10:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"846a5fe4ec924b838496d94ef1c5d36bedad3a77","revisions":{"846a5fe4ec924b838496d94ef1c5d36bedad3a77":{"kind":"REWORK","_number":1,"created":"2022-08-04 16:40:38.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/95/852195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/95/852195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/95/852195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/95/852195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/95/852195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/95/852195/1"}}},"commit":{"parents":[{"commit":"165c21ef99c62e3ad32c350403eb9b732cb16e04","subject":"Fix CI issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/165c21ef99c62e3ad32c350403eb9b732cb16e04"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-08-04 16:40:20.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-08-04 16:40:20.000000000","tz":180},"subject":"Suppress deprecation warning from docutils","message":"Suppress deprecation warning from docutils\n\nChange-Id: I5904b75b0c60d8fd99e4fb57b752109b960cff45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/846a5fe4ec924b838496d94ef1c5d36bedad3a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/846a5fe4ec924b838496d94ef1c5d36bedad3a77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[]}
