)]}'
{"id":"openstack%2Frally~863844","triplet_id":"openstack%2Frally~master~I861996a8e6663d2a11ed7f1b3a23b7ebe0747eb2","project":"openstack/rally","branch":"master","topic":"drop-python-dev-from-bindep","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-01-27 15:53:13.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-05-19 18:33:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I861996a8e6663d2a11ed7f1b3a23b7ebe0747eb2","subject":"Remove python-dev from bindep and tox4 fixes","status":"ABANDONED","created":"2022-11-07 10:04:47.000000000","updated":"2023-05-19 18:33:37.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"297f8f036c01516601098159531ed6fdbf597812","_number":863844,"virtual_id_number":863844,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-27 15:46:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-07 10:29:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-21 18:53:22.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2023-01-27 15:24:03.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"3a0dadf9379c6f3b5ac3db8353dd0736e7cb5181","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-07 10:04:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0b83aa1c99180587e22f4c87f87d23e6078fa9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 10:29:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef809f253298432fa343d986faef4496\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe6f5056f254a4baf2b88cc62cbcc48 : SUCCESS in 4m 07s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/cbe9c270ee1e48cc87c02afb4d0b1e0d : FAILURE in 2m 52s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ae3742e120a440bfac150078867fd389 : SUCCESS in 7m 09s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/927f9b5cd02a40cc8c2c3cc873d91d6c : SUCCESS in 3m 16s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/b67170e27ae24dd5916129ce51c71949 : SUCCESS in 3m 16s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/e92853f428684d51907d21fba5cb7d8e : SUCCESS in 4m 19s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/5e0501c7649f4d99aa0f2415e8362664 : SUCCESS in 3m 16s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/214f04a3bb734ea58a26a11a3842f806 : SUCCESS in 3m 11s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/636bad1d516f44c991cd16240ded41d8 : SUCCESS in 3m 10s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/1bbbddd06b9e441f90a537dbc483cac6 : SUCCESS in 6m 06s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b2b14f24420e4195a4af79dbfce51862 : FAILURE in 6m 52s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/d2173313dce740319f1151e78930fa6e : SUCCESS in 6m 29s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fb87c0572c64797869904b492b827ec : SUCCESS in 6m 10s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/181275fffc0f404aa1b3eed614f9e296 : SUCCESS in 5m 55s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/0628cfda95a84221a08f1a3ee9b8cac9 : SUCCESS in 7m 10s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/d800af06920840e88db6580fe3077969 : SUCCESS in 7m 45s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/0fa964e9945d43098228524a7eb8a297 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1387a92416e019e870fe125da8b880ef07674a55","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2022-12-21 18:53:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"102c89a353568f5d1ce5ce558af5cb712606fd67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 19:17:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/706b4c748673445cb9b9f815a57e9143\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0072ccd817c4661bbfd05526ae94cc4 : FAILURE in 4m 00s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/d2b0a56cd20f46b5b4e9e20bc26e6955 : FAILURE in 3m 13s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/df8b403f828c4313871ce03d0e66c51b : SUCCESS in 3m 56s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/64616fecf5354b20957c7922d18dc4a9 : SUCCESS in 3m 14s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/0fbcda17e95b4e1bb3894453f2f7c1ad : SUCCESS in 4m 06s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/4d37f514d5674f0cb481950ceb9ebbd7 : FAILURE in 2m 44s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/655ebec03bde40d6adcdc07a421793d4 : FAILURE in 2m 36s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/45555391663a4c54a46b2f0b14962581 : FAILURE in 2m 54s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/ec51a6bcef2f49d8be90a92d31934802 : SUCCESS in 3m 01s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/99af00b7aed843ffb97d7a76d804765e : SUCCESS in 5m 37s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/42753419eb4a4cad96af95de15754591 : FAILURE in 2m 49s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/c598c67754ca4ee9b331a7399c1e41dc : SUCCESS in 6m 22s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7996b5ac4c940d097a6f6f778eb6c0b : SUCCESS in 6m 14s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f04542a09575473db7b157b2a405c2e4 : SUCCESS in 5m 55s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/8016177cd372456bb04c8495eb80fbcf : SUCCESS in 7m 10s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/4100e8bf3f774186a899f3485e041c85 : SUCCESS in 7m 21s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/ed6accea98574d9e82cd4be04cf6ebdf : SUCCESS in 8m 33s","accounts_in_message":[],"_revision_number":1},{"id":"124d22122f58f53012bdf19859d382b2753f8183","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-27 15:24:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2abf968360214e7029efa1a3df47ef6ed0f1b721","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-27 15:29:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3c664aa9a50744dbcbac7ec2ee778f4525385449","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 15:46:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0d42ddd16c6440caef323a648b801c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e97963907a24db299ad1de3d2adf4b0 : FAILURE in 4m 00s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/93b42cbc00c34a77a14461ff6d5f1f6f : FAILURE in 3m 00s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/64316f32888b40e7a602d62341ecbb83 : SUCCESS in 4m 57s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c0b45ec1ef1491c9257c9e699e5225e : SUCCESS in 4m 03s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/99f926a4b39c41549dd1b1bbcfba58d5 : SUCCESS in 3m 28s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/5a44a45fd51348cb9706774e2ed4ab78 : SUCCESS in 4m 13s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/0020704ec873476aa120163bd5300b87 : SUCCESS in 3m 09s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/c336ff8d9cef4ffcb0b95ad8c46790bd : SUCCESS in 4m 05s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/aa75936144cc41219d2e79ecc9b59266 : SUCCESS in 3m 38s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/8499f095470840bdba1cf49cb6da0e78 : FAILURE in 4m 31s\n- rally-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8911296087a141e28b8cadb74e1401c0 : FAILURE in 4m 38s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/46ebf576523444c7818c683b58aed06b : SUCCESS in 6m 41s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f14ba5fe077462f82afc82f7e481377 : SUCCESS in 7m 27s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/636c008c6e2b4b3fa2ed042602f49fbc : SUCCESS in 5m 53s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/41510010c58742e2ace64f32bb673e5a : SUCCESS in 8m 14s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/5edc38432b5c49b3ba5968db7284297e : FAILURE in 7m 22s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/f2b794b1c29c4f79bfcf35e279c3f8e1 : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":2},{"id":"d61a52f16754abd92a16ae60ee529039f43065a2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-27 15:53:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bbdea8febcceb0f4aa3f4ccfa6fe7e2215d7770c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-30 12:16:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"297f8f036c01516601098159531ed6fdbf597812","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 18:33:37.000000000","message":"Abandoned\n\nmerged as a part of another change","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4ef069b782204595376c67f0ff494dc146727cf2","revisions":{"91c859dd011e28ae9e3bb16f3e2b0f83a473557c":{"kind":"REWORK","_number":1,"created":"2022-11-07 10:04:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/44/863844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/44/863844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/44/863844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/44/863844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/44/863844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/44/863844/1"}}},"commit":{"parents":[{"commit":"03155a82e2510112e029b668962eb4434524af9f","subject":"remove unicode literal from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/03155a82e2510112e029b668962eb4434524af9f"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-07 10:04:42.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-07 10:04:42.000000000","tz":60},"subject":"Remove python-dev from bindep","message":"Remove python-dev from bindep\n\nIt is no longer supported by jammy and lead us to the following errors with the announce-release job.\n\n```\nNo package matching \u0027python-dev\u0027 is available\n```\n\nChange-Id: I861996a8e6663d2a11ed7f1b3a23b7ebe0747eb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/91c859dd011e28ae9e3bb16f3e2b0f83a473557c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/91c859dd011e28ae9e3bb16f3e2b0f83a473557c"}]},"branch":"refs/heads/master"},"4ef069b782204595376c67f0ff494dc146727cf2":{"kind":"REWORK","_number":2,"created":"2023-01-27 15:29:07.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/44/863844/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/44/863844/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/44/863844/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/44/863844/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/44/863844/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/44/863844/2"}}},"commit":{"parents":[{"commit":"03155a82e2510112e029b668962eb4434524af9f","subject":"remove unicode literal from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/03155a82e2510112e029b668962eb4434524af9f"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-07 10:04:42.000000000","tz":60},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-01-27 15:27:26.000000000","tz":330},"subject":"Remove python-dev from bindep and tox4 fixes","message":"Remove python-dev from bindep and tox4 fixes\n\nIt is no longer supported by jammy and lead us to the\nfollowing errors with the announce-release and docs\njobs.\n\n```\nNo package matching \u0027python-dev\u0027 is available\n```\n\nAlso Update tox.ini to be tox4 compatible.\n\nChange-Id: I861996a8e6663d2a11ed7f1b3a23b7ebe0747eb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4ef069b782204595376c67f0ff494dc146727cf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4ef069b782204595376c67f0ff494dc146727cf2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
