)]}'
{"id":"openstack%2Frally~883652","triplet_id":"openstack%2Frally~master~I77eba97596df4448065982956c3b6fb08c8e45db","project":"openstack/rally","branch":"master","topic":"ci","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2023-05-19 20:24:14.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"I77eba97596df4448065982956c3b6fb08c8e45db","subject":"Keep CI up-to-date","status":"MERGED","created":"2023-05-19 10:09:45.000000000","updated":"2023-05-19 20:35:12.000000000","submitted":"2023-05-19 20:35:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883652-ci","meta_rev_id":"747b9b290c6bdc0c5a8b51324e4a83f3d8a93d8a","_number":883652,"virtual_id_number":883652,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-19 20:35:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2023-05-19 20:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2023-05-19 20:24:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-19 11:10:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c37ce7d9896f6c87cf2f761fd59e52af1dada62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 10:09:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ea30c24162bd4ace079c0662e8a4c9d955d6892","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 10:47:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61d6972078442d9a3221aae09b96a15b01b2f5cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 11:10:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3222eb7335d24d3a93560e08e127f96b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7c2d226a7ed34fa98e658e3ff873af39 : SUCCESS in 1m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d41d98d0d3947cc887f1e3e9e18b475 : SUCCESS in 4m 07s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/14bc4c1a827a429f886a0a3015d8edce : SUCCESS in 3m 27s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/10bdad19a87f4767ace3a29268a1ad81 : RETRY_LIMIT in 1m 48s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/39a5fabc2a404515a1a10ea46751190e : RETRY_LIMIT in 1m 31s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/ff29f41ff5a84367a58f1212e2e304a5 : RETRY_LIMIT in 2m 18s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/5900557a1c634e1193022944cb88479c : RETRY_LIMIT in 2m 49s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/d74456d413574e4d87a6e90caf22f9a6 : RETRY_LIMIT in 2m 15s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/c298bee043d14ab39684ae45fb51d2e0 : RETRY_LIMIT in 1m 41s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/d017ed0e66804e02b321573fa5eaf821 : RETRY_LIMIT in 1m 47s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/b062dc1e560047a6b7f6fa2b690f6f69 : RETRY_LIMIT in 1m 38s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/c9816690f9d14d47950e9c15d6925914 : RETRY_LIMIT in 2m 18s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ef0ebb4a8de448e4b7ae7f37a1251085 : RETRY_LIMIT in 3m 25s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06c5234aa2514ed78c28110f3c94c695 : SUCCESS in 6m 33s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b6ef30f2ba446f090a024dacbfa33ec : RETRY_LIMIT in 1m 29s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/84ebfc8e1c1b4c718b61210b6e3bed1d : SUCCESS in 6m 55s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/fe98dd9e06b84a02bd65daf3a3d6a3b6 : RETRY_LIMIT in 2m 39s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/b2527e754c4741e79e9f912f409c9e24 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":2},{"id":"ee8b1c0a98401149fd91f6a5c481896ca087aaa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 11:49:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4e4fc39ed21e587b937d12247b19e54d96baf75a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 12:28:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29f891aadcba40dca603e7ce893f88cc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/04c2e7fc3661472398fed28fc54adef7 : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/561644f4a3cc40eda451fc7d5beeeae8 : SUCCESS in 4m 48s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b65bc83b76864bcb9c60f2f3bc1d5b24 : SUCCESS in 3m 20s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/9284c3203d614ca3a8f2aa51e788d279 : RETRY_LIMIT in 1m 40s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eb2f7b4f89141cca217f4630ead5ae1 : RETRY_LIMIT in 1m 42s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/48b8f33ba2184111b89a0e210a706e73 : RETRY_LIMIT in 2m 19s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/76bf29a8b93240b2b7506f90dfb6d8d6 : RETRY_LIMIT in 1m 46s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/15e29220c1cc4eac9f871fc29538754c : RETRY_LIMIT in 1m 35s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/acf0d3b6b6ff40a5ac0d87d0d9d6a6fd : RETRY_LIMIT in 1m 46s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/ca0c1f1001864a30bc0cd992c47faf21 : RETRY_LIMIT in 1m 44s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/87c6d6870017438ea330818f20de7221 : RETRY_LIMIT in 1m 28s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/fe7993745aef4289bb95fdc70e7c77df : RETRY_LIMIT in 2m 15s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/c119a9387de34395b28fcd038c9ab22e : RETRY_LIMIT in 1m 48s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4c414491bc647d19f1ef642e16c97b4 : SUCCESS in 5m 41s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a13a8f677f7b45b8a6dcfa09fd263708 : RETRY_LIMIT in 1m 27s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/01566a4b08e04e43a39056c22ddf0af1 : SUCCESS in 6m 57s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/7dfc3caa01e84219b20e1b84b26cdf6a : RETRY_LIMIT in 1m 54s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/049e1d3dfb79473fb01abcc440db40ea : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":3},{"id":"441c5701aa0ace78f35bbbf27c161248e13c3b81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 12:58:24.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1c5d8872e44c15ce9f96f85db3608459ea5c90bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 13:41:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/030cabde6e5747d18dfbab3537af445f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d9cfe6847c5549a2ac181c9569049f4d : SUCCESS in 1m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/836bb489596d481e841aafe053301f4f : SUCCESS in 4m 53s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/0e0d247238c74e61ae4ec9f6384c3601 : SUCCESS in 4m 27s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/d427af7480c7441ea956d36fde7cb0a9 : RETRY_LIMIT in 1m 42s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca0cb69e25264558b31dc9110ef053ae : RETRY_LIMIT in 1m 41s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/bd174a0d53774cc3a4e369f2b1090d02 : RETRY_LIMIT in 1m 38s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/8f83ee4f5dca4b1a92069cb087c50cb3 : RETRY_LIMIT in 1m 28s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/7a6c3d669a4640c385f2ef91e9a40f76 : RETRY_LIMIT in 1m 45s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/b830697540494a958e60a073dd8c8090 : RETRY_LIMIT in 1m 44s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/c0c7f2371f544252ac3e1ca638105dd4 : RETRY_LIMIT in 1m 25s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/fdf7ca2f2c5c4631bb3de71c5d1cd4ec : RETRY_LIMIT in 1m 40s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/9121898150ae427da6ddb6cff7744180 : RETRY_LIMIT in 1m 41s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/9ae25ca2c1bf4700a727b88d4f5803b0 : RETRY_LIMIT in 2m 35s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40dbe4b253cf44acbdd74644c25d28e9 : SUCCESS in 5m 53s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d72fc07bc8d4e13a52f1dfe98c25635 : RETRY_LIMIT in 1m 29s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/9eef4b0033844aab86be3b771168af02 : SUCCESS in 8m 09s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/e88b781ab86e43329b1a53f0ebc76509 : RETRY_LIMIT in 1m 53s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/01a81686f15d4c9caa9d61ef36176494 : FAILURE in 7m 36s","accounts_in_message":[],"_revision_number":4},{"id":"bfe5a91fc99bc9cc50849c66aee1303047c483c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 14:51:50.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e4faaff7a5a572ed0dfc2e181cfb1a4e4ae56693","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 15:10:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/648008ca91454367994ecd6a51234183\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7188acbc6c2e40e48d3b2f65dc5d8aa6 : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dcecaa2c7d844949de018788046789c : SUCCESS in 5m 04s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/504dc82f6c0646928113ea49aee62397 : SUCCESS in 3m 40s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/e678f163866b4422ba028e397bf83b65 : RETRY_LIMIT in 1m 53s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9925260f69b46d9962955e31005429b : FAILURE in 2m 59s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/1c0b494be9c440e781f2ecd2456c7bc2 : RETRY_LIMIT in 2m 51s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/90153b5da13246888f1ae5551ac3bafb : RETRY_LIMIT in 2m 28s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/b9597fc16d0343da8da998edd834f8c5 : SUCCESS in 4m 06s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/fdd5f417d68144a18a2d98f8b259da48 : SUCCESS in 3m 18s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/28aea960e30a4078892cb56684acc0a0 : FAILURE in 3m 32s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/b885a1c4751341b39a5624226cefc08c : FAILURE in 3m 45s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/da66ee4f8da7459da6ddbd8c96c9753b : FAILURE in 4m 34s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/821de29df04940f7863677af808a4dae : FAILURE in 3m 28s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a190fce5bff431b98c665e47339cf40 : SUCCESS in 5m 56s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2cd4512cafd4b41ba5ef61b51c84016 : RETRY_LIMIT in 1m 29s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/edd0621e9bbb4b81aaaaaeee06a6aa9a : SUCCESS in 7m 38s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/c89bde69d37742cdaaf20fdce6a05307 : RETRY_LIMIT in 2m 06s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/a67ba2e962d94b1e9738124060a708c5 : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":5},{"id":"617bee0c4ba1fecab84a667fac7d9c5c85313914","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 15:24:06.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"b4949a9d388db6dcef35719b5753a0804a897926","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 15:40:39.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f4cbd53bcc8ceae3cf12e383927aad72449c5916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 16:03:24.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/09170683b45e4bf2910f3f714256ffb8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/73337444306b4f64a0bedc7d1b336c0e : SUCCESS in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5763a53e0c0243609e0df61e81e810e2 : SUCCESS in 4m 06s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/7bc516fc9d2b4690a2ed013cf5137404 : SUCCESS in 4m 28s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f8bdde0abe24404b83c44628707268ca : SUCCESS in 3m 27s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eafc4e6846847fcb35e6cfe9d8458fd : FAILURE in 3m 18s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/aaca1dfe068643d38e20274671e9a2ca : SUCCESS in 2m 57s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/e45764238bdb48b39235148ca2d65ec8 : SUCCESS in 4m 43s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/810dfb71a7fe4c3a98cb59239670fdc6 : SUCCESS in 3m 04s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/13377fe119464a03871dcb77099ac046 : SUCCESS in 3m 20s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/cf83675c35e749168b58f015759321df : FAILURE in 2m 59s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/18fd55ec654e4d739e7d04b6f64649c3 : SUCCESS in 2m 46s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/67a2c65e1f7e4f11ace7e5272006a636 : SUCCESS in 7m 43s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/f328b69f041f4c25a66257e4547400ac : SUCCESS in 6m 23s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31c3b9789d0641f79435df9a78e466f9 : SUCCESS in 5m 50s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8edc4986a049463f8292b585eea855a7 : RETRY_LIMIT in 1m 31s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/83931aa146d240b48d26dc2527d31e72 : SUCCESS in 7m 40s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/39bc43e084c94d899e12019046c3a959 : RETRY_LIMIT in 1m 56s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/14463db0b6b64b44a6083a83cab19736 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":7},{"id":"eb3a0516d5fbbb4f0f867c627aef9854b639a591","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 16:28:49.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"1b7af8c5233d7c84a8334468a33b1d624bb95aaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 16:49:07.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d07b29e5445b4ee5b4ec2c6456e0297e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4b682fa1d114444b8870f9caf02499ba : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4600eeb161f9468b9294185275d8f629 : SUCCESS in 4m 35s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/ee39558ebf84410097c7f75b673ee673 : SUCCESS in 4m 29s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/2ebe3ca58d4d483c9cc272b3791a4f60 : SUCCESS in 4m 23s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/54e26cc6db274de3a785b26e5722cd62 : FAILURE in 2m 55s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/980706e187084ccfb008801a157616c2 : FAILURE in 3m 33s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/558488cde51547b08ac1790c5b661a84 : FAILURE in 3m 22s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/a7ca8b80ef784139a17999bfe781acd6 : FAILURE in 3m 32s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/dee4bba535604cd384c62e88bcc6351a : FAILURE in 3m 23s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/ce65ab516c4b48e78f3cb79243bce5cd : FAILURE in 2m 54s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/ebca848811b644aaa54bd54db11cefcd : SUCCESS in 2m 47s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/0e48deee8ac6476a84b72cbff269c3b3 : SUCCESS in 5m 30s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/601319985ce6413f842d27d5375c55d2 : SUCCESS in 6m 57s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cccdba26da744e88e99d2415d76282a : SUCCESS in 5m 50s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0071e651ccc438184ab9b84f9c4d859 : RETRY_LIMIT in 1m 52s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/d8d1b976c7874b578d4d0668a415cb96 : SUCCESS in 6m 37s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/28aa3f59e1af4cacb28ac1ed4f69a6d6 : RETRY_LIMIT in 2m 03s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/e67c12204d9b452ba203cff58e567ec8 : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":8},{"id":"e51ec12bc90f7f85763a3579581e99b10c108d2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 17:16:00.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"2b3b7ea7aaa614ec1af3266c7a59c58925a2d412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 17:29:28.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"77d2dc494d9111d3c77f281ca3094746cd3745ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 17:43:42.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9b29195ddfdf3fa8682fb20774eb84198be2fa78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 17:53:45.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"59abd86fe48916072792d42da51760f6bc25b3d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 18:12:28.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"bef88a319182216ed3c6728b8104c34439d99c06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 18:22:18.000000000","message":"Uploaded patch set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"ba01bc2c0dae400455891f1effa276ab9acabca7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 18:41:51.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"a482aef91a846bf66f8dca5e05cd45a2537cac4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 18:42:02.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9d89660c6eb43448cb88d3ec314bd6c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8b3e2a69f9ea43f19c66aa538cb48bdc : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3dc071d8d34900821b98d505f43935 : SUCCESS in 7m 11s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/ffc9d362de06456082d922204f9d1f5d : SUCCESS in 4m 05s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/65d577bcee9d4528b5d3f9d67c4b03fd : SUCCESS in 4m 57s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/18d09a7d434c4dde98d3f7ce060fb35a : SUCCESS in 3m 00s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/37e734f943844e1bb33ccfe92941faa4 : SUCCESS in 3m 42s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/44ea65bc3c654636b72276978b7a25b3 : SUCCESS in 3m 43s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/ba90f30c3cc14732ae872b05416e050b : SUCCESS in 3m 05s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/7bf02e93b58e47d8b3949ce526e097e7 : SUCCESS in 3m 04s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/15e8aa5c62ec4c7aabe519fb9f95222f : SUCCESS in 2m 59s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/1a04334cc833486b83d2e4f19e29687c : SUCCESS in 2m 37s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/b2508e48a33441ed8f187b6f20eac5b1 : SUCCESS in 5m 12s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/37ee643c1cca48a1afa6d001f426ddf4 : SUCCESS in 6m 09s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a01757f8c6c24de4965957eb8eb28a62 : SUCCESS in 5m 33s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c9ad0d42a1342f7a3999671811180bd : SUCCESS in 5m 32s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/2ce9a90ffd6e4c69bf1a957e8fd07226 : SUCCESS in 7m 16s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/c24409a11a35495da8d1cfcf53e62b1d : SUCCESS in 6m 18s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/05af5e370ce245f995c5c7763bfd9e64 : SUCCESS in 8m 27s","accounts_in_message":[],"_revision_number":14},{"id":"181d38371c1407335630b437e760ed397e0780aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 18:55:32.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ac2a010ce774666a5d88c5627e8e95d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/de1854490d7742859ae7b244074c26f7 : SUCCESS in 1m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bf30d78ce6546809eec21369b9453d7 : SUCCESS in 4m 49s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/4bf46949db864bbd8dd1456aa3bb9622 : SUCCESS in 4m 14s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/89772e0edd4f4fa3a1a8153fd27eb323 : SUCCESS in 5m 13s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f59db9213cb46a39b4eede4cac3ef54 : SUCCESS in 3m 01s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/b6d1bdb0759f4f6fb202fc7f9b558463 : SUCCESS in 3m 47s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/4a8b86a5f5644566b6882946616003e2 : SUCCESS in 3m 33s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/310bc6b1cbfc4f9bb2483199a3cc980e : SUCCESS in 3m 08s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/5de8794120d04bff896cc38ab4082987 : SUCCESS in 3m 12s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/21acc26902364595adc0ce10a8e5b7e0 : SUCCESS in 3m 16s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/52bd7dec94a74824b34d91ba95feb184 : SUCCESS in 3m 06s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/d8fc70ac21494e359d3b1d8ce136781d : SUCCESS in 5m 11s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/221833f8412b4f1d801b3605807b4061 : SUCCESS in 6m 07s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b102b55d76446c8868bafe5ca225438 : SUCCESS in 5m 02s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b04eb6682684cac8d281f476b5c4d5b : SUCCESS in 5m 10s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/2d20976570244deeac8b716ebb925074 : SUCCESS in 6m 47s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/430bf278d6a34acc86e0a2dbe56d19fe : SUCCESS in 5m 46s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/4772b814f9e84b16a42c8087b58bbdd9 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":15},{"id":"559f80423f3025da01bd8017a1b6405b6b989839","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 20:24:14.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"ff391f11eff68c8cd363f6eebb6207f2e04f4a67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 20:24:41.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"a2d994870de20795dc6a03d76eb9433ef60c7438","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 20:35:12.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04d938bcbc584958b9673118e0516922\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e26b4d5b77e44df9ae7f1eb92c25fc69 : SUCCESS in 1m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e38c4cdb4dd4915bccbf5f36e07cc94 : SUCCESS in 4m 00s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/6a64d8fc77d64281b23c78c5903b5636 : SUCCESS in 3m 53s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/924f3edd1bca4b158491fcb5386f974a : SUCCESS in 3m 48s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/5541ebeb15b646b49e102f83603cbfe4 : SUCCESS in 3m 02s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/120e1a3d831542f7a6059766b39d164e : SUCCESS in 3m 11s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/a9ebdeff2354440cae89d897436b5f29 : SUCCESS in 3m 35s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/eaa5149f9edb456997ef2fe8c4f83e39 : SUCCESS in 3m 17s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/4961dcba5d4b4de8bff555b59fa5dcf5 : SUCCESS in 3m 18s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/2c55125e0ba448239f544eeaaedc4ca5 : SUCCESS in 2m 56s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/f995d79c27954e03858210b11141e4c8 : SUCCESS in 5m 16s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/3701593499af4d22b9fe8858bc78bfbb : SUCCESS in 5m 58s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f90652fef25640f09d0d8ee4718def72 : SUCCESS in 5m 25s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c01f27ed83649feb1035b461d5c58b8 : SUCCESS in 5m 25s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/41b9760017bc4f9c9b1658f454cef3b6 : SUCCESS in 6m 33s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/8de0965801764adc9da2700c4c14a53c : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":15},{"id":"747b9b290c6bdc0c5a8b51324e4a83f3d8a93d8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 20:35:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"1694ddc50eb5e3dd872232d5ec96ebd3914c4100","revisions":{"49d6ae2b0013ade628d8aa1e3b68ba6d337fad1b":{"kind":"REWORK","_number":1,"created":"2023-05-19 10:09:45.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/1"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"subject":"WIP: update ci","message":"WIP: update ci\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/49d6ae2b0013ade628d8aa1e3b68ba6d337fad1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/49d6ae2b0013ade628d8aa1e3b68ba6d337fad1b"}]},"branch":"refs/heads/master"},"be52117dbb7efe227e6a62dde4f8273523af5cb3":{"kind":"REWORK","_number":2,"created":"2023-05-19 10:47:53.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/2"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:47:39.000000000","tz":120},"subject":"WIP: update ci","message":"WIP: update ci\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/be52117dbb7efe227e6a62dde4f8273523af5cb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/be52117dbb7efe227e6a62dde4f8273523af5cb3"}]},"branch":"refs/heads/master"},"9f740753065145b0efd759cb8602e7145afee4a6":{"kind":"REWORK","_number":3,"created":"2023-05-19 11:49:57.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/3"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 11:49:48.000000000","tz":120},"subject":"WIP: update ci","message":"WIP: update ci\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9f740753065145b0efd759cb8602e7145afee4a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9f740753065145b0efd759cb8602e7145afee4a6"}]},"branch":"refs/heads/master"},"31dcf4ed49e898f16d2541540495c56a1d21cb48":{"kind":"REWORK","_number":4,"created":"2023-05-19 12:58:24.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/4"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 12:58:14.000000000","tz":120},"subject":"WIP: update ci","message":"WIP: update ci\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/31dcf4ed49e898f16d2541540495c56a1d21cb48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/31dcf4ed49e898f16d2541540495c56a1d21cb48"}]},"branch":"refs/heads/master"},"541e203b0223f9ae2b72f5e68d8d1c46527ecd0e":{"kind":"REWORK","_number":5,"created":"2023-05-19 14:51:50.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/5"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 14:51:35.000000000","tz":120},"subject":"WIP: update ci","message":"WIP: update ci\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/541e203b0223f9ae2b72f5e68d8d1c46527ecd0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/541e203b0223f9ae2b72f5e68d8d1c46527ecd0e"}]},"branch":"refs/heads/master"},"506b0f9ffb932351aa74d269dee31e0e96d5858c":{"kind":"REWORK","_number":6,"created":"2023-05-19 15:24:06.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/6"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 15:23:58.000000000","tz":120},"subject":"WIP: update ci","message":"WIP: update ci\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/506b0f9ffb932351aa74d269dee31e0e96d5858c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/506b0f9ffb932351aa74d269dee31e0e96d5858c"}]},"branch":"refs/heads/master"},"8486d1c4b1efe1a24c3987b57bf30ee47b5ee1f2":{"kind":"REWORK","_number":7,"created":"2023-05-19 15:40:39.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/7"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 15:40:29.000000000","tz":120},"subject":"WIP: update ci","message":"WIP: update ci\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8486d1c4b1efe1a24c3987b57bf30ee47b5ee1f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8486d1c4b1efe1a24c3987b57bf30ee47b5ee1f2"}]},"branch":"refs/heads/master"},"afaa715d90130555e3504372e02fc25d1ebd37ec":{"kind":"REWORK","_number":8,"created":"2023-05-19 16:28:49.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/8"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 16:28:38.000000000","tz":120},"subject":"WIP: update ci","message":"WIP: update ci\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/afaa715d90130555e3504372e02fc25d1ebd37ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/afaa715d90130555e3504372e02fc25d1ebd37ec"}]},"branch":"refs/heads/master"},"455ce557cebd5ac484ff20e4a20ca49b13d8576e":{"kind":"REWORK","_number":9,"created":"2023-05-19 17:16:00.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/9"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 17:15:56.000000000","tz":120},"subject":"WIP: update ci","message":"WIP: update ci\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/455ce557cebd5ac484ff20e4a20ca49b13d8576e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/455ce557cebd5ac484ff20e4a20ca49b13d8576e"}]},"branch":"refs/heads/master"},"2f439c3f82f15f4a905645c434b9bc5630f2b5e3":{"kind":"REWORK","_number":10,"created":"2023-05-19 17:29:28.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/10"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 17:29:20.000000000","tz":120},"subject":"WIP: update ci","message":"WIP: update ci\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2f439c3f82f15f4a905645c434b9bc5630f2b5e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2f439c3f82f15f4a905645c434b9bc5630f2b5e3"}]},"branch":"refs/heads/master"},"ca2e6c0c25ebc0ec1168f6539d6f9451549c057d":{"kind":"REWORK","_number":11,"created":"2023-05-19 17:43:42.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/11"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 17:43:36.000000000","tz":120},"subject":"WIP: update ci","message":"WIP: update ci\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ca2e6c0c25ebc0ec1168f6539d6f9451549c057d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ca2e6c0c25ebc0ec1168f6539d6f9451549c057d"}]},"branch":"refs/heads/master"},"03b75e94c9a714824bb0fb23ffb461737114fc24":{"kind":"REWORK","_number":12,"created":"2023-05-19 17:53:45.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/12"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 17:53:36.000000000","tz":120},"subject":"WIP: update ci","message":"WIP: update ci\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/03b75e94c9a714824bb0fb23ffb461737114fc24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/03b75e94c9a714824bb0fb23ffb461737114fc24"}]},"branch":"refs/heads/master"},"7a9ef2b982bc7ccf61f764064cb4c2b7893f6fb0":{"kind":"REWORK","_number":13,"created":"2023-05-19 18:12:28.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/13"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 18:12:21.000000000","tz":120},"subject":"WIP: update ci","message":"WIP: update ci\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7a9ef2b982bc7ccf61f764064cb4c2b7893f6fb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7a9ef2b982bc7ccf61f764064cb4c2b7893f6fb0"}]},"branch":"refs/heads/master"},"5cf610fbc5055efc6bc1407ae35900f5da4017cf":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2023-05-19 18:22:18.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/14"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 18:14:40.000000000","tz":120},"subject":"WIP: Keep CI up-to-date","message":"WIP: Keep CI up-to-date\n\n* Fix CI job that check rally installation with CentOS 9. Previously it\n  used CentOS 8 node :)\n* Start using Ubuntu 22.04 LTS as much as possible\n* Use python 3.10 by default for most tox envs\n* Fix issues with new hacking\u0026flake8\n* Rewrite the way of preparation for tox\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5cf610fbc5055efc6bc1407ae35900f5da4017cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5cf610fbc5055efc6bc1407ae35900f5da4017cf"}]},"branch":"refs/heads/master"},"1694ddc50eb5e3dd872232d5ec96ebd3914c4100":{"kind":"REWORK","_number":15,"created":"2023-05-19 18:41:51.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/883652/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/883652/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/883652/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/883652/15"}}},"commit":{"parents":[{"commit":"7e5a996027e16c0734101c7ce408a68205148bdc","subject":"Re-enable tox cover zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5a996027e16c0734101c7ce408a68205148bdc"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 10:09:39.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 18:37:56.000000000","tz":120},"subject":"Keep CI up-to-date","message":"Keep CI up-to-date\n\n* Fix CI job that check rally installation with CentOS 9. Previously it\n  used CentOS 8 node :)\n* Start using Ubuntu 22.04 LTS as much as possible\n* Use python 3.10 by default for most tox envs\n* Fix issues with new hacking\u0026flake8\n* Rewrite the way of preparation for tox\n\nChange-Id: I77eba97596df4448065982956c3b6fb08c8e45db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1694ddc50eb5e3dd872232d5ec96ebd3914c4100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1694ddc50eb5e3dd872232d5ec96ebd3914c4100"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
