)]}'
{"id":"openstack%2Frally~883690","triplet_id":"openstack%2Frally~master~I27b1f108584892d64b78366d2bc39f616583af1e","project":"openstack/rally","branch":"master","topic":"docker-python","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2023-05-20 07:26:15.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"I27b1f108584892d64b78366d2bc39f616583af1e","subject":"Use official python docker images","status":"MERGED","created":"2023-05-19 18:22:18.000000000","updated":"2023-05-20 07:58:13.000000000","submitted":"2023-05-20 07:58:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883690-docker-python","meta_rev_id":"6b4def844285a7e945cea1e2b7a7b89c7759b8c3","_number":883690,"virtual_id_number":883690,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-20 07:58:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2023-05-20 07:26:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2023-05-20 07:26:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-19 18:45:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c1f11ba3447d4016e25a3de7e095d37eebfc693","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 18:22:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a6c9a2b8d0c60dfe61ff02b0d270cec735f801d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 18:26:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c801c943821284b06d287c75f8571d1a02233937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 18:45:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/440d89a8a4ef4115a668c62c8c99aeb7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d6875f1e6c0419c98df26a51568f17b : SUCCESS in 5m 31s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/d24cc0f8521b494b968d590b92b2b2ed : SUCCESS in 4m 10s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f58352b47c194407acbb679122108403 : SUCCESS in 3m 58s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d2a36a941e941579e885bd088029f46 : SUCCESS in 3m 29s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/691d9430e5fa4ba199228875e2b364b6 : SUCCESS in 3m 11s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/fb464c408b1d4254ae725edfed1a85d1 : SUCCESS in 3m 36s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/8c687bdcd2514023bdfbb7d2fbfae299 : SUCCESS in 3m 13s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/b2591dd32c1349f0b53b5f61eb74e03d : SUCCESS in 3m 03s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/67d9eb571b6c490897e8e445afcab49d : SUCCESS in 3m 09s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/15b4e1ce1e8e4ed5954cfbc09a1d157d : SUCCESS in 3m 04s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/4686b6d1d8894b94a248f6ff169614c3 : SUCCESS in 6m 13s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/21fc2e5c31784cdfaddf62f27bf1f7ca : SUCCESS in 6m 37s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f34406e0838e41639eb3fda554f49dfc : SUCCESS in 5m 27s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f881bead583d43ad89faf1bebc9711e5 : SUCCESS in 5m 37s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/e5316a401cc54dfb944d8c5d40a796f8 : SUCCESS in 6m 57s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/3fe447d4d5c546d593efc4db9d5b236f : SUCCESS in 6m 05s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/b8f00f79bbde45289d81589518991542 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":2},{"id":"0ecb01a3587112dc3185e53eb57da9e5c19b258c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 21:30:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"74705534fcd99036f8eaa460d302c7ebba6c64b1","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 21:31:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"71ef1d056055082161a7a63365bf685cb7d11dee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 21:45:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b391c13214d4904afed41a169f73e4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c359169f7b23492ba29cc82a15c1ac14 : SUCCESS in 5m 30s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/9277edb944434ed2a257b6b374142c3f : SUCCESS in 4m 45s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/c56552d589c64657a5ee109794b04f23 : SUCCESS in 4m 14s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/35141afb16424c7b9faa7eae433f221c : SUCCESS in 2m 43s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/3cf603746941418fabe8e60b4b5698a1 : SUCCESS in 3m 50s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/45075b7a438f48feb7451aa355332d45 : SUCCESS in 4m 41s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/0075aa3e2c984a10a9d7fdef3bf1b166 : SUCCESS in 3m 07s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/acf165cd8f274b1abb89710c6c806de3 : SUCCESS in 3m 12s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/24c197705e3e4a61a96c60a00d5880c0 : SUCCESS in 3m 11s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/7ddaf30348484094bd50bc05a47b4569 : SUCCESS in 3m 15s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/03ee12ead4ab4cfeb246aacbe16ae7bf : SUCCESS in 6m 20s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/fe170512e19b449ca2b4c534e941ffd4 : SUCCESS in 6m 12s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/583c21e48248446db5e78458abc70da8 : SUCCESS in 5m 24s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/562e28716fcc491597244af4657ec4dd : SUCCESS in 5m 47s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/46e3aa0808dd45ef9364a4c205221392 : SUCCESS in 7m 32s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/b7158877a7954557ba27af357c0f036c : SUCCESS in 6m 22s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/90b54b636370412f953acc1de898188c : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":3},{"id":"55fa8ca4ee942f988614cef54243a299f0bde9d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 21:46:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"de167d2944108ff741b541339877e0ebab168ae0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 21:50:53.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b08b267e64fcfac3cf2f10c678c843447a51a028","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 21:53:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d60e2db1bbcf417b093affdb6c8d392ea0e4c05d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 22:04:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc3832cbbe6d4951874dab2ef193da2e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36e351950064fb6839e39324a2dc5e4 : SUCCESS in 4m 44s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/d81cec5822934560b55dead3de6a19d7 : SUCCESS in 3m 58s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/1cca761d58ba49c6bd77f3e3734774ac : SUCCESS in 4m 25s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/351042a317104593a426e6237960fb4e : SUCCESS in 2m 55s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/3cf0112171924600a07eb07c785a28ed : FAILURE in 4m 04s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/271b491d02cf49fb91918414d7017641 : FAILURE in 3m 23s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/977110dc0ffb445aa9f0d710c2217576 : FAILURE in 2m 47s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/6a04284166d8497391ff43d77d2cd2a8 : FAILURE in 3m 15s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/a475a5f420de43cb9809c7aa2f3c8ccd : FAILURE in 2m 54s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/304b858fde0c4feb8f627a9430471df9 : SUCCESS in 3m 04s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/4085a1ee3a76424baa52daefee0bf312 : SUCCESS in 5m 01s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/1fa8d2c5a911424bbfa9ee13d48ecd7f : SUCCESS in 6m 19s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b104a12b703e4ee0816fda691ed4e939 : SUCCESS in 5m 21s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/823ec6bc3df04cb5a4643d9030584839 : SUCCESS in 5m 22s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/89ced520aa814946831d93be9b831aaa : SUCCESS in 7m 01s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/106064cc9dd74f9daa221e137dc599f0 : SUCCESS in 5m 52s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/bb79ca4ea97142dfb6b0bdcbb33a941c : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":4},{"id":"8797ebea0c82c8bd53bc4c92d004f9517082f4ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 22:23:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8039af34d7b12444b58e9de92a8ebf72d13464e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 22:35:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b38743a48654aa28e58165acc79a70c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf66a557cb443f5a83ff3663a512067 : SUCCESS in 5m 12s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/3c4a5d5b7ab247a2a8d93af5c887d367 : SUCCESS in 5m 07s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ec70c4cb11544bd09e60bc22039fc805 : SUCCESS in 4m 12s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec1a2c92502d4885a81565195961e0e6 : SUCCESS in 3m 26s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/2b77dfe04bcb414894492f1f972dca87 : SUCCESS in 3m 03s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/6fb0365aeed74f12a05e35f7247bb7b8 : SUCCESS in 4m 43s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/72fe2dc7c95f4c4da63a82f001d1fb62 : SUCCESS in 3m 29s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/f9f94c0053c04d4d9e663931561e57a6 : SUCCESS in 3m 16s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/7fa5a28aa45846b38fe5ef845521801a : SUCCESS in 2m 57s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/d5b1aae71fb84956a280298d8b597665 : SUCCESS in 3m 29s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/34ae7e4a94864c7c8579a1d0b2249649 : SUCCESS in 6m 00s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/f3bd43af052842f4b569facc72da2dcc : SUCCESS in 6m 04s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cacf27ffaf3b40889e3363f25bda4b0c : SUCCESS in 6m 02s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd073c85739f4fa8859cad240705e44e : SUCCESS in 5m 29s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/9625ba4440264b37ae11ec29d4371955 : SUCCESS in 6m 17s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/78367f69a2c749e088b24dd467779cd6 : SUCCESS in 6m 00s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/36272e56406743b09d33ee5790093dfc : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":5},{"id":"c4161c9aaaee68804de577d3516642324e66658a","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-20 07:26:15.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f3c054f9702b30bdfac8bb06e6a21736d5a0f9c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-20 07:26:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"66506352c8564c3ac485966e6087716ac0d09c7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-20 07:58:13.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fd9a0ff23d64d7ab39bd19161832195\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da81e1f8756742948a0c63937fedf836 : SUCCESS in 5m 20s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/ae10ec2b119544178baf89fca7ceff60 : SUCCESS in 4m 11s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/fa3a95981ff44bf58d64563e5476aaff : SUCCESS in 3m 55s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/4157bc21c6364a2eb119b7cdf35b1291 : SUCCESS in 3m 17s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/2911c4e4cf774df6b75f588251758e7b : SUCCESS in 3m 27s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/abbf8d540abe43aea329cc053dae8220 : SUCCESS in 6m 04s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/009e18d251914867a899177c725557ab : SUCCESS in 3m 31s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/164ffd4572a942e098771c9cb624fd99 : SUCCESS in 5m 37s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/3e5832a9546c4ed09d47f13acdf34fed : SUCCESS in 3m 23s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/23d7e1bf12084854a64017a4bcbda787 : SUCCESS in 5m 37s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/9e0aaeecbc8e46eeac431f511f42dcf1 : SUCCESS in 5m 50s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c22846489a54bcc9165fd477d9d4dda : SUCCESS in 5m 52s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8921cb327c9e4dd68168fc664989b9a0 : SUCCESS in 5m 27s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/b7c7b4681dd745388911a89d41c05d6e : SUCCESS in 6m 27s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/60fe3f9c33364c9ba9ecb00b036f19fd : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":5},{"id":"6b4def844285a7e945cea1e2b7a7b89c7759b8c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-20 07:58:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"177ad8b5fb1ac6b422ea2b0cc7878f43f7b3d8fd","revisions":{"3735fb361b024439e2c3b081836f0c3b11db97f6":{"kind":"REWORK","_number":1,"created":"2023-05-19 18:22:18.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/90/883690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/883690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/883690/1"}}},"commit":{"parents":[{"commit":"5cf610fbc5055efc6bc1407ae35900f5da4017cf","subject":"WIP: Keep CI up-to-date","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5cf610fbc5055efc6bc1407ae35900f5da4017cf"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 18:22:01.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 18:22:01.000000000","tz":120},"subject":"WIP Use official python docker images","message":"WIP Use official python docker images\n\nChange-Id: I27b1f108584892d64b78366d2bc39f616583af1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3735fb361b024439e2c3b081836f0c3b11db97f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3735fb361b024439e2c3b081836f0c3b11db97f6"}]},"branch":"refs/heads/master"},"0c82013d94461df3b18a0b985f02b3939648847b":{"kind":"REWORK","_number":2,"created":"2023-05-19 18:26:48.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/90/883690/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/883690/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/883690/2"}}},"commit":{"parents":[{"commit":"5cf610fbc5055efc6bc1407ae35900f5da4017cf","subject":"WIP: Keep CI up-to-date","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5cf610fbc5055efc6bc1407ae35900f5da4017cf"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 18:22:01.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 18:26:34.000000000","tz":120},"subject":"WIP Use official python docker images","message":"WIP Use official python docker images\n\nChange-Id: I27b1f108584892d64b78366d2bc39f616583af1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0c82013d94461df3b18a0b985f02b3939648847b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0c82013d94461df3b18a0b985f02b3939648847b"}]},"branch":"refs/heads/master"},"f38b3d717e1cec149bd126968528cf5c419c1a0f":{"kind":"REWORK","_number":3,"created":"2023-05-19 21:30:57.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/90/883690/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/883690/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/883690/3"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 18:22:01.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:30:16.000000000","tz":120},"subject":"Use official python docker images","message":"Use official python docker images\n\nChange-Id: I27b1f108584892d64b78366d2bc39f616583af1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f38b3d717e1cec149bd126968528cf5c419c1a0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f38b3d717e1cec149bd126968528cf5c419c1a0f"}]},"branch":"refs/heads/master"},"d28d7e216ca9d6cac82f78f1dc39e712b2316eae":{"kind":"REWORK","_number":4,"created":"2023-05-19 21:50:53.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/90/883690/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/883690/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/883690/4"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 18:22:01.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:50:48.000000000","tz":120},"subject":"Use official python docker images","message":"Use official python docker images\n\nChange-Id: I27b1f108584892d64b78366d2bc39f616583af1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d28d7e216ca9d6cac82f78f1dc39e712b2316eae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d28d7e216ca9d6cac82f78f1dc39e712b2316eae"}]},"branch":"refs/heads/master"},"177ad8b5fb1ac6b422ea2b0cc7878f43f7b3d8fd":{"kind":"REWORK","_number":5,"created":"2023-05-19 22:23:11.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/90/883690/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/883690/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/883690/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/883690/5"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 18:22:01.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 22:23:04.000000000","tz":120},"subject":"Use official python docker images","message":"Use official python docker images\n\nChange-Id: I27b1f108584892d64b78366d2bc39f616583af1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/177ad8b5fb1ac6b422ea2b0cc7878f43f7b3d8fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/177ad8b5fb1ac6b422ea2b0cc7878f43f7b3d8fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
