)]}'
{"id":"openstack%2Frally~883709","triplet_id":"openstack%2Frally~master~I933d3ffdbb48f6393d0d0665ad83ff166824250d","project":"openstack/rally","branch":"master","topic":"update-uc","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2023-05-23 07:35:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"I933d3ffdbb48f6393d0d0665ad83ff166824250d","subject":"Update requirements and upper-constraints.txt","status":"MERGED","created":"2023-05-19 21:50:22.000000000","updated":"2023-05-23 07:50:52.000000000","submitted":"2023-05-23 07:50:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883709-update-uc","meta_rev_id":"aad12a92897c096660898c153fe69f27554a324c","_number":883709,"virtual_id_number":883709,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-23 07:50:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2023-05-23 07:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2023-05-23 07:35:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-19 22:28:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8bd3f066af749c124a56d024158c9a293dacfb25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-19 21:50:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"262cce8dc36c78c495d22540f52786389e263862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 22:28:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2077f55aa2c497e9049fe27cd60884d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ae24344252493a99c57b814d4fbc49 : SUCCESS in 4m 06s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/28e91bfbd81b45f4923be6bccdc5444c : SUCCESS in 3m 47s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/b817709af7e54c05913f682e6cf2bc30 : SUCCESS in 3m 56s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/4af7c96d3d1d44b0ad55312037c610a3 : SUCCESS in 3m 00s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/3add6f8a145045eaa69f75f470433ab4 : FAILURE in 2m 24s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/c58d5a0621e34db9ae59566e7334c15d : FAILURE in 2m 40s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/47d4af0bfa2b45d4b47115258cd8b973 : FAILURE in 4m 23s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/81564f1f24c0462d881e30824df0a712 : FAILURE in 3m 29s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/9847601f246044c19c742f6c745d7ade : FAILURE in 3m 11s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/2c152b5c169c47ceb2b97c1962bae991 : SUCCESS in 3m 17s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/36a6b6eea88a4f12bd0504c5deb27ba3 : SUCCESS in 5m 58s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/1dfa8dec4b594be080f391e0970041a0 : SUCCESS in 5m 58s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f83772a1bdb544cbb04b45cffbb7e341 : SUCCESS in 5m 18s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ec95a65ae774a0b8bcf32e0277ebed8 : SUCCESS in 5m 25s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/44117fb2eb2644f5a02ce1b3c9c691ce : FAILURE in 2m 22s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/05814d3d35964a94b07c61d63b0ba037 : SUCCESS in 6m 10s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/e55c74e972d947ad8646029c782d8c6f : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b1e69fb323bac4d239fece5523877fa8c7dca512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-20 08:03:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"27d7503148a865750129ba87d5f4a403d2faf0c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-20 08:40:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ff3ae9c57b1470eb8a3bb0183f595ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b761c009caa4789b0d8ef2b1220777e : SUCCESS in 5m 16s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/bbf9070cf441444da4cc45d976e08556 : SUCCESS in 4m 19s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/44446c7914e647b9ab85f8fdd4206738 : SUCCESS in 3m 38s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/627b056037d546138849a8ff3cc0fb3b : SUCCESS in 3m 28s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/21396af4161d44dfb43146d33e8d1aea : FAILURE in 2m 20s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/2c5fa96f12164b1d8652f83d4fab4f9d : FAILURE in 2m 19s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/a86da2d3d49a4ac487f45b60b6e98809 : FAILURE in 3m 02s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/54e8d99912ce456ca538ad6eae736be9 : FAILURE in 2m 53s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/a0cec81b68bb42ed97fb5be96b97b00b : FAILURE in 3m 12s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/183526e2b65540a2a13843d59b9b1eac : SUCCESS in 2m 58s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/e282649085894cb49c4099aafac627c5 : SUCCESS in 5m 15s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/4af974887b6e40d9bad300f4aaa497c4 : SUCCESS in 7m 15s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4585fbf09440414bb37d1aff2d2abbda : SUCCESS in 8m 10s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8dffb4cd34b4f568cbda4417e014764 : SUCCESS in 5m 16s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/45a11dc7968f4e1bb50143d342f15c82 : FAILURE in 2m 24s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/14781c91ee114741a1edf12778c46043 : SUCCESS in 5m 57s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/cbb6b7345b5d4468a8eabe11e7111b61 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":2},{"id":"5f54f1bd3515727881a87b94f79cfd0385998391","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-21 07:30:39.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"07491dcc898077ae1c41b1b456e8d0da9cf05692","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-21 07:42:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5ae5ebe41d0bd2b43f9755794a27ad1bbc3fbe0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-21 08:16:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf1f782a888b48afa698d6a54425752c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87251f1408a747f9963124ec89d3941c : SUCCESS in 4m 14s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/fa97fc95d52c4c67b1b8030340d682c8 : SUCCESS in 4m 12s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/1d39ee504d294cc39bd4eeabfed77b64 : SUCCESS in 4m 01s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d5e789ddc624e5c84ed5b0c9c7e316e : SUCCESS in 2m 39s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/738bc1c4aae6499a8be0548b20de869b : FAILURE in 2m 46s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/f14e9c65a303435fb750c957cf625726 : FAILURE in 2m 24s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/2b287fbb6c1f4d39a6777458e3a478a9 : FAILURE in 3m 07s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/3b82fa419e4d403d9d4b6876380aec07 : FAILURE in 3m 19s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/f2cb2e03c2c64e13b99550d1d879c8cf : FAILURE in 3m 20s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/4b4d16c1add54ad1b741db020b0e24d0 : SUCCESS in 2m 51s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/0f21153a682b407296b03bfdbe2f596d : SUCCESS in 5m 26s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/3079fea9a27b4bf8a539dbd52f667aac : SUCCESS in 6m 12s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d05b20a0bce4489aff35fcb53a97927 : SUCCESS in 5m 31s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04b1ae97ace64610a8fe11b3e78fb707 : SUCCESS in 5m 28s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/2bc8d17a784d456e986bd8cfd599ee2d : FAILURE in 2m 44s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/e9029ebc9ca34d0abaa052dae52f7c86 : SUCCESS in 5m 50s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/e2755c00af4849a880f512fbb16e633f : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":4},{"id":"fe69394e207a4ac364494100b9bdafdc046f2a7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-21 08:43:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"40845d0f6f9f3b1b8112a21aecf13c011e57316e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-21 08:53:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/81858252b51345909e2348e824b1d516\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d950647f66fb4ca79cd28706c8d323f4 : SUCCESS in 4m 20s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/85a21552d37b4d67b005dc83b8ee4103 : SUCCESS in 4m 27s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/de02c629b4f4418c98bdcc01a766f6d5 : SUCCESS in 4m 32s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0b176f7801240f5bb75190c03856f97 : SUCCESS in 2m 48s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/e50007c5f3bc450cbcba530d0440e9b3 : FAILURE in 2m 01s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/787f7680c2464449b63220d189aff28d : FAILURE in 3m 19s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/76eca5c9fe2c4a979eccceb6210655f3 : FAILURE in 3m 25s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/e6121ff58b814226ba933fe44b001285 : FAILURE in 3m 34s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/f7c3e62598e34df6a91465a9fa34ea52 : FAILURE in 3m 01s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/83cad9e2705345ada94fd03bf08b25dc : SUCCESS in 2m 42s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/9e8598f97ca3427b8c5efb4c162bcbc6 : SUCCESS in 5m 35s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/c46073a85be04646adadcd0377ce3d38 : SUCCESS in 6m 32s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/607b4212b9874afc80528d92fbc23585 : SUCCESS in 5m 35s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74791f4454394330bb4f434d2b2a2af0 : SUCCESS in 7m 41s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/e251d7053b144a36a146b5c8e0da2b05 : FAILURE in 3m 20s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/51238a780ff8405c94d32597076eba3e : SUCCESS in 7m 31s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/8bb57b1556a04e2e88b59d57940482c1 : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":5},{"id":"6e6d4a3a05f471bc72de5e86a3a16692bbe6f51c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 18:18:12.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"16ec6d5a5b7b5d30d202766e24d642a08d10e755","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 18:35:27.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3d547cdb367e653b7dc59450523249f96acd9a02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 18:45:34.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"29fe16efdadc00a962b8f6dc2e953baad7b1a198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 19:02:30.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/376c387bf66846f698544e3936be0def\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bfbdfed8c9642d8b2e196a28bb80577 : SUCCESS in 5m 05s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/9bd40eaf70884d628dc5e5c7ecda34b3 : SUCCESS in 4m 18s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/a88fa0f2d4bd4feeac9b36ba8e3a9fb3 : SUCCESS in 3m 36s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9fc77396284b6ca2867012521dd599 : SUCCESS in 2m 39s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/158d55ae9bd44f4ea968f97a2fdfa495 : FAILURE in 4m 28s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/08378217fd93445892a68080d9a7de32 : FAILURE in 4m 10s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/33ee5b2f93b2475e9de5ce1840bd86d2 : FAILURE in 3m 16s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/efc8766873b3450ebc56da2e79c60558 : FAILURE in 3m 20s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/d3cd0fe79cb540a4bd34167fc8a2dd17 : FAILURE in 3m 42s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/9cd38ba531ad47ada3dbe9e6892e3280 : SUCCESS in 3m 19s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/aa64c66af1e04b708682d64dff9a867b : SUCCESS in 6m 00s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/1bc673d0ba5c41089d53160ddb7df5f2 : SUCCESS in 7m 04s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf194c5862b3495aa27cb76cc69dd460 : SUCCESS in 5m 30s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57c97b20a881420b8df321ff8fcba4c6 : SUCCESS in 7m 44s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/2c36f20a70e240e6b281ea86c39fa5a7 : FAILURE in 2m 38s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/531933f3eb6445439289ba7854637b47 : SUCCESS in 6m 05s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/7d2bf7d2c04e4a0fba2b4a3720e1c38c : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":8},{"id":"1e6bc32bff61419a4aed0e6f6b384becdd288438","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 19:02:46.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"8b0e6adc1b2d1caf7c1c7d55eaf8a1627a30602e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 19:15:37.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d8ebbd81e96ad9f07f0a17c7156b412f8bb1a843","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 19:33:59.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"7e5ea8dae0fb1ba04c402922d641aef7598db367","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 19:40:35.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ebf6ee2341689ea707df52551d800fe4d42ab99c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 19:48:55.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"35e081d65de5dbed9aa8980c1e6fb3fc35793393","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 19:57:41.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"bb8532a83b43dfed08ecec4cf7c4a9433d5e4e1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 20:03:29.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"863110bebff583fbfbfbfe8785c90e93d7f1e7a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 20:08:43.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"8521a33d25f727dee9981b33c3a7df9f5d9168d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 20:13:42.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"f0fa59b5278f54801703c7d2d221f4260314d63d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 20:19:07.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"ded1a8cbbc8a78ed54d67f2f7f36ba85f47c947b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 20:26:59.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"4423672eab1c8fa23c81670dcdac72b9bc2d10ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 20:33:52.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"6d673f24f87ccc6c2b3a43a49836765275b0652d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 20:42:43.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"5c7c3659f5001452431e472a75192d170e038e77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 20:43:10.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"0deb30b53e620ca352ca1f8044729679d01b1270","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 20:49:46.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"2fbf106232a388ced5e18a7e8c9e584e38da2de5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 21:01:24.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20e67cabe542466f820b5f4d9aaf6ad5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad8b277e1fca42879b21cf65796a320c : SUCCESS in 4m 48s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/37aa043c9f66413ea3c0ce26ad0e52d4 : SUCCESS in 3m 29s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/d46f5109acaa48889ac4545773607fcc : SUCCESS in 3m 54s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/11ada9cabf874fe2a71f6a5493daa59d : SUCCESS in 2m 57s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/1c887d5405974f60ba1bd05a8b805aa2 : SUCCESS in 3m 36s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/4b2fe18db32d4958a282c4ab3eb351ad : SUCCESS in 3m 34s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/38925198003048c9835fa643d688acb7 : SUCCESS in 3m 40s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/9251b4d1f93a4d6b941351ae41588db3 : SUCCESS in 3m 37s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/dc54f20518c4482b92acb54317b7c45d : SUCCESS in 3m 06s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/2493c35d6f3c4a2fa2abd024f385cb1f : SUCCESS in 2m 58s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/9b576bf9a8ec4a3baf6e7814cc8ad137 : SUCCESS in 5m 41s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/a68b3e8adc924cc2a50dcdc616592330 : SUCCESS in 6m 11s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fce7bfbbc6549e09d1e032cdd567f7e : SUCCESS in 5m 29s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d462afd5f3dd4f31b0dfedb07e0bf32c : SUCCESS in 5m 12s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/74b5aae8af3d484e9488f304ce554739 : SUCCESS in 6m 56s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/f675e38ab8614f62b881122837e09762 : SUCCESS in 6m 05s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/e8ddf09e570347099e9ca4d9c3f46462 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":23},{"id":"9b395ad2cdd3e2b65799b00265784fff097a181d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 21:02:11.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"f62abcdb37587b588ec88e60135562ea4ddcf996","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-22 21:02:32.000000000","message":"Uploaded patch set 25: Commit message was updated.","accounts_in_message":[],"_revision_number":25},{"id":"d5e800ba4071a758eff1e21f6e6001d056872ea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 21:23:26.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b159030019aa40b2a086874d27b71c0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/383ae6c4906f4e80aa008c87c578b575 : SUCCESS in 4m 43s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e9b63f0487e847499a29ac4c797c2570 : SUCCESS in 3m 51s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/5dd377310efa460782d18fe1a177fdf4 : SUCCESS in 4m 01s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/10fa7d0f847f4e31a7422993f021367e : SUCCESS in 2m 40s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/32c541747a9842a8816b097bb9b292da : SUCCESS in 3m 32s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/7ca01b4b41c7450cac940650f03f4223 : SUCCESS in 3m 37s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/dea2c14d56e54715b517ca7700e5df36 : SUCCESS in 2m 49s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/d9250949aaac4dc68831c76ccdaff69d : SUCCESS in 2m 57s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/7467a8f60ae24edea885216869e2ab8b : SUCCESS in 3m 00s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/7f8d942b6bc34fa8b19ae156a3c42bd2 : SUCCESS in 2m 54s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/bdb714c187a8428eb997c4a0340b2043 : SUCCESS in 5m 27s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/141e6566cb1c46b5b576f9721636fc3a : SUCCESS in 6m 12s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61ffca455fb2415fbac6f4665c3b095d : SUCCESS in 5m 33s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/488ad1902a1b4084a3de9369ed35f7e3 : SUCCESS in 6m 15s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/30023d313daf49e6b7b9e5f84b9f27ad : SUCCESS in 6m 26s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/0ba88f84b1214c25b38ead6635f248d2 : SUCCESS in 5m 59s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/2780b6e4931b40bfbb72b2464fe29d0d : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":25},{"id":"f3fe6952263923613876a1d0a2583ef84ff867fe","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-05-23 07:35:24.000000000","message":"Patch Set 25: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":25},{"id":"a383a17415c913d83c0644ff363c2c666ef42aea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 07:35:43.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"ca2e0073b31ebbe82526a57e94ea01c119c1ffa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 07:50:52.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9141277c562141acb12442e981d9c5e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6503f0811ab4e4ba9652fce96764bdf : SUCCESS in 4m 32s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/45eff14747a0431d8743ed07824d3985 : SUCCESS in 4m 02s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/1747969f5e2747de9a3ca7c2b3909825 : SUCCESS in 3m 42s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/260c1b1ae81644858431a1e5f28cf69c : SUCCESS in 2m 37s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/ea74f87a397c417084c67c98e088586d : SUCCESS in 3m 50s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/5323458c0a0543a9a59dd64492f68de1 : SUCCESS in 3m 26s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/cb6991e81fae4d96a6da4cfdce5dcfe0 : SUCCESS in 3m 18s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/33b7fd9815fb4270a6d1ccbbd37083aa : SUCCESS in 3m 22s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/b3db6f3985c045a5ba6d7621df047588 : SUCCESS in 3m 11s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/540a53e48a21464195a0679f1bee45fc : SUCCESS in 5m 52s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/7b35f9d12c854d5284b21636229d87f1 : SUCCESS in 6m 27s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35d63034a70141df9440d6968dc3a931 : SUCCESS in 5m 25s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13b4b78d488443628f12e3a979ac91f3 : SUCCESS in 5m 11s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/073d8503e8864bf281523e5c9ba7aede : SUCCESS in 6m 39s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/c08a8e8f189d41cc9e5caee869ecdc40 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":25},{"id":"aad12a92897c096660898c153fe69f27554a324c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 07:50:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"b08ea0e9379bcee6f81715b937ff8471a57db1df","revisions":{"ae4033ecd0ba7a99e6949448231d2820c9b4309f":{"kind":"REWORK","_number":1,"created":"2023-05-19 21:50:22.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/1"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:50:10.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ae4033ecd0ba7a99e6949448231d2820c9b4309f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ae4033ecd0ba7a99e6949448231d2820c9b4309f"}]},"branch":"refs/heads/master"},"41407b4ff00fa5463560370c91ce2a3f47ba9c3d":{"kind":"REWORK","_number":2,"created":"2023-05-20 08:03:00.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/2"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-20 08:02:48.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/41407b4ff00fa5463560370c91ce2a3f47ba9c3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/41407b4ff00fa5463560370c91ce2a3f47ba9c3d"}]},"branch":"refs/heads/master"},"1e1b08ff4a692fe3a60f414a59fc1ab31c049658":{"kind":"REWORK","_number":3,"created":"2023-05-21 07:30:39.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/3"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-21 07:30:26.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1e1b08ff4a692fe3a60f414a59fc1ab31c049658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1e1b08ff4a692fe3a60f414a59fc1ab31c049658"}]},"branch":"refs/heads/master"},"c2bffe19c41574e9c464c3fababa79b3c8937a9b":{"kind":"REWORK","_number":4,"created":"2023-05-21 07:42:35.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/4"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-21 07:42:30.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c2bffe19c41574e9c464c3fababa79b3c8937a9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c2bffe19c41574e9c464c3fababa79b3c8937a9b"}]},"branch":"refs/heads/master"},"af8dfdf2de680b9fe522ffd65561985c76d1820b":{"kind":"REWORK","_number":5,"created":"2023-05-21 08:43:15.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/5"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-21 08:43:07.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/af8dfdf2de680b9fe522ffd65561985c76d1820b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/af8dfdf2de680b9fe522ffd65561985c76d1820b"}]},"branch":"refs/heads/master"},"401653adb3cff9e890be6a7fd36c672750dc34ff":{"kind":"REWORK","_number":6,"created":"2023-05-22 18:18:12.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/6"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 18:18:02.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/401653adb3cff9e890be6a7fd36c672750dc34ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/401653adb3cff9e890be6a7fd36c672750dc34ff"}]},"branch":"refs/heads/master"},"5864e726b5b08e2f72c3e411d33f5c51e2244150":{"kind":"REWORK","_number":7,"created":"2023-05-22 18:35:27.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/7"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 18:35:22.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5864e726b5b08e2f72c3e411d33f5c51e2244150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5864e726b5b08e2f72c3e411d33f5c51e2244150"}]},"branch":"refs/heads/master"},"0550e9d2f0d50fdb59d791cdf906f02b43be0346":{"kind":"REWORK","_number":8,"created":"2023-05-22 18:45:34.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/8"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 18:45:30.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0550e9d2f0d50fdb59d791cdf906f02b43be0346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0550e9d2f0d50fdb59d791cdf906f02b43be0346"}]},"branch":"refs/heads/master"},"f1ba3dea90906dbec5480cd992b9ca587dcabbaa":{"kind":"REWORK","_number":9,"created":"2023-05-22 19:02:46.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/9"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 19:02:28.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f1ba3dea90906dbec5480cd992b9ca587dcabbaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f1ba3dea90906dbec5480cd992b9ca587dcabbaa"}]},"branch":"refs/heads/master"},"d6359dec84ae58b0e4e13f9808b658840b162aa5":{"kind":"REWORK","_number":10,"created":"2023-05-22 19:15:37.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/10"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 19:15:32.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d6359dec84ae58b0e4e13f9808b658840b162aa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d6359dec84ae58b0e4e13f9808b658840b162aa5"}]},"branch":"refs/heads/master"},"143d2cf30ebc8f1cb6959256b96c14f01951d7e0":{"kind":"REWORK","_number":11,"created":"2023-05-22 19:33:59.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/11"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 19:33:51.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/143d2cf30ebc8f1cb6959256b96c14f01951d7e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/143d2cf30ebc8f1cb6959256b96c14f01951d7e0"}]},"branch":"refs/heads/master"},"d625d612a1ef26d40ff6306ff69b5b853764d0cd":{"kind":"REWORK","_number":12,"created":"2023-05-22 19:40:35.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/12"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 19:40:28.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d625d612a1ef26d40ff6306ff69b5b853764d0cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d625d612a1ef26d40ff6306ff69b5b853764d0cd"}]},"branch":"refs/heads/master"},"c63466e4959bc2689bd986b822b84d0afa5ae9ed":{"kind":"REWORK","_number":13,"created":"2023-05-22 19:48:55.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/13"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 19:48:50.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c63466e4959bc2689bd986b822b84d0afa5ae9ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c63466e4959bc2689bd986b822b84d0afa5ae9ed"}]},"branch":"refs/heads/master"},"0a586c560592658983209a90aad31a024f3a3e2f":{"kind":"REWORK","_number":14,"created":"2023-05-22 19:57:41.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/14"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 19:57:36.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0a586c560592658983209a90aad31a024f3a3e2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0a586c560592658983209a90aad31a024f3a3e2f"}]},"branch":"refs/heads/master"},"194989b9bd2383ef60afdbdb05365dd0c7c20f6a":{"kind":"REWORK","_number":15,"created":"2023-05-22 20:03:29.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/15"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 20:03:23.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/194989b9bd2383ef60afdbdb05365dd0c7c20f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/194989b9bd2383ef60afdbdb05365dd0c7c20f6a"}]},"branch":"refs/heads/master"},"dd51d2563490b6bde50fad42a3618a1324d2563b":{"kind":"REWORK","_number":16,"created":"2023-05-22 20:08:43.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/16"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 20:08:38.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/dd51d2563490b6bde50fad42a3618a1324d2563b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/dd51d2563490b6bde50fad42a3618a1324d2563b"}]},"branch":"refs/heads/master"},"2f804214c98526243b0585f6f0e9f4440ef05a45":{"kind":"REWORK","_number":17,"created":"2023-05-22 20:13:42.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/17"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 20:13:04.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2f804214c98526243b0585f6f0e9f4440ef05a45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2f804214c98526243b0585f6f0e9f4440ef05a45"}]},"branch":"refs/heads/master"},"5d14295edcce8a536bf8afe07ff8460b0de7be8d":{"kind":"REWORK","_number":18,"created":"2023-05-22 20:19:07.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/18"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 20:19:01.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5d14295edcce8a536bf8afe07ff8460b0de7be8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5d14295edcce8a536bf8afe07ff8460b0de7be8d"}]},"branch":"refs/heads/master"},"f73b693b78e1a00788e1bc5931b16ef2e0179c4c":{"kind":"REWORK","_number":19,"created":"2023-05-22 20:26:59.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/19"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 20:24:30.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f73b693b78e1a00788e1bc5931b16ef2e0179c4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f73b693b78e1a00788e1bc5931b16ef2e0179c4c"}]},"branch":"refs/heads/master"},"1ddceaff1fc278d3bb9d5c81eb0b15077523bc70":{"kind":"REWORK","_number":20,"created":"2023-05-22 20:33:52.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/20"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 20:33:34.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1ddceaff1fc278d3bb9d5c81eb0b15077523bc70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1ddceaff1fc278d3bb9d5c81eb0b15077523bc70"}]},"branch":"refs/heads/master"},"7f4afc4ef3cb0f34499a5d6b88d83f9fae5d80cb":{"kind":"REWORK","_number":21,"created":"2023-05-22 20:42:43.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/21"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 20:42:25.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7f4afc4ef3cb0f34499a5d6b88d83f9fae5d80cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7f4afc4ef3cb0f34499a5d6b88d83f9fae5d80cb"}]},"branch":"refs/heads/master"},"e92310faeea356cdb0bb177ad6107f787616f6d6":{"kind":"REWORK","_number":22,"created":"2023-05-22 20:43:10.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/22"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 20:43:06.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e92310faeea356cdb0bb177ad6107f787616f6d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e92310faeea356cdb0bb177ad6107f787616f6d6"}]},"branch":"refs/heads/master"},"14809773f90c1b4b44a552ca6bfe81f50de1d17a":{"kind":"REWORK","_number":23,"created":"2023-05-22 20:49:46.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/23"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 20:49:41.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/14809773f90c1b4b44a552ca6bfe81f50de1d17a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/14809773f90c1b4b44a552ca6bfe81f50de1d17a"}]},"branch":"refs/heads/master"},"4937d6621bee331629d64a00a07caf0af9d464e8":{"kind":"REWORK","_number":24,"created":"2023-05-22 21:02:11.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/24"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 21:01:18.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nIn addition, this commit includes soome fixes to make migrations work\nwith swqlalchemy 1.4.0\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4937d6621bee331629d64a00a07caf0af9d464e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4937d6621bee331629d64a00a07caf0af9d464e8"}]},"branch":"refs/heads/master"},"b08ea0e9379bcee6f81715b937ff8471a57db1df":{"kind":"NO_CODE_CHANGE","_number":25,"created":"2023-05-22 21:02:32.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/09/883709/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/09/883709/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/09/883709/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/09/883709/25"}}},"commit":{"parents":[{"commit":"e65935a938d4af2279b31f1efcd5dec8b67a0295","subject":"Merge \"Keep CI up-to-date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e65935a938d4af2279b31f1efcd5dec8b67a0295"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-19 21:48:19.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-05-22 21:02:21.000000000","tz":120},"subject":"Update requirements and upper-constraints.txt","message":"Update requirements and upper-constraints.txt\n\nIn addition, this commit includes soome fixes to make migrations work\nwith sqlalchemy 1.4.0\n\nChange-Id: I933d3ffdbb48f6393d0d0665ad83ff166824250d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b08ea0e9379bcee6f81715b937ff8471a57db1df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b08ea0e9379bcee6f81715b937ff8471a57db1df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
