)]}'
{"id":"openstack%2Frally~901541","triplet_id":"openstack%2Frally~master~Ic16ce1d9b479dd53e1e157699dd80594318e77fd","project":"openstack/rally","branch":"master","topic":"py36-37","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2023-11-21 18:44:08.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"Ic16ce1d9b479dd53e1e157699dd80594318e77fd","subject":"Drop support for py36 and py37","status":"MERGED","created":"2023-11-21 11:52:05.000000000","updated":"2023-11-21 19:12:23.000000000","submitted":"2023-11-21 19:12:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901541-py36-37","meta_rev_id":"d0b1065aa22eb2a05d7c7b19ec277cf9546f042c","_number":901541,"virtual_id_number":901541,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-21 19:12:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2023-11-21 12:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2023-11-21 12:11:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-21 11:52:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f17aa0a971416f123a3e04b95ba6c27943f47dfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-11-21 11:52:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6907990e5b6daac2814253570ce652e7ed875d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 11:52:38.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/rally on branch master.  The\nproblem was:\n\n  Job rally-tox-py311 not defined\n\nThe problem appears in the the \"opendev.org/openstack/rally\" project stanza:\n\n  project:\n      templates:\n        - docs-on-readthedocs\n      vars:\n        rtd_webhook_id: \"52691\"\n      check:\n        jobs:\n          - rally-tox-cover\n          - rally-tox-docs\n  ...\n\n  in \"openstack/rally/.zuul.d/zuul.yaml@master\", line 2","accounts_in_message":[],"_revision_number":1},{"id":"81f745955b71a5984d2b41245868432f53f50057","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-11-21 11:57:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"85f6d9af13fe90e91482ade3048e05c32705a731","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-11-21 11:58:08.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"59d2aaf9e19c33b7039cb674dce22f74d449d6c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 11:58:08.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/rally-openstack on branch master.  The\nproblem was:\n\n  Job rally-tox-py36 not defined\n\nThe problem appears in the the \"opendev.org/openstack/rally-openstack\" project stanza:\n\n  project:\n      check:\n        jobs:\n          - rally-tox-cover\n          - rally-tox-pep8\n          - rally-tox-py36\n          - rally-tox-py37\n          - rally-tox-py38\n          - rally-tox-py39\n  ...\n\n  in \"openstack/rally-openstack/.zuul.d/zuul.yaml@master\", line 1","accounts_in_message":[],"_revision_number":2},{"id":"aadcbf00d8169161159e3be6d443152b21d5d9e0","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-11-21 12:11:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"506f3edbec738f4c5003178750b9547e50ccdf39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 12:17:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a67343af9dc43a28e6d8683873865f0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dd5a87ccfa0b4e46b90be2b15eec96b9 : SUCCESS in 1m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14cb633b9f6c44b8a4057a3245e16725 : SUCCESS in 4m 26s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/31dca70216cf49fdb1f737601fa68890 : SUCCESS in 4m 38s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/746ef4785ba34ca2a1790b7e24326bff : SUCCESS in 4m 30s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/78ebdefa647d4861bbe25316f22a1057 : SUCCESS in 2m 55s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/8af82f80416f45229848ce49aae740f6 : SUCCESS in 3m 17s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/8c14adfebfb54ca7a96cb5ccbfa36354 : SUCCESS in 3m 34s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/81589eade89e45df862afc8a6778cd39 : SUCCESS in 3m 28s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/8618296b6303488cb568627059b26e47 : SUCCESS in 3m 41s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/4d87b511fb6245d3ac9e89cb9e5572c2 : SUCCESS in 3m 58s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/f0ffa9c475534f22a7eb89c2f2d3583e : SUCCESS in 5m 25s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/dfd20182f1eb41e99299774726aea0e2 : SUCCESS in 6m 24s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a855a09b1adb4f05a1dd4f61449ccc7a : SUCCESS in 6m 00s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f129dfad2ba34115a98355e23841b3c6 : SUCCESS in 6m 12s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/8c7500b6e3c6486d9d4b8bfdc51cbe0d : SUCCESS in 7m 08s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/a9ca1d7d5dd6425f81a3684ab9319a6f : SUCCESS in 6m 09s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/8a358661530344a681176cebfc31426c : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":3},{"id":"d402982cd72dcd033f8311e1b4e0d522594b536e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 12:17:59.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f81dc4e4143d42a6b055438a9dfeffe3\n\n\nWarning:\n  Change 892104 in project openstack/rally-openstack does not share a change queue with 901541 in project openstack/rally","accounts_in_message":[],"_revision_number":3},{"id":"34f2416aad4d732042aa2cbe8e159e7b7d86e61b","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-11-21 18:44:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"983f84dfe158bc2dc20b423c7141d49550a692e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 18:57:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7dbb60b1f294e5092d569a5ce0d524e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/35f7d21a7bbf407ab4ccff8158810a29 : SUCCESS in 2m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56df30dc29e14806b58f57ae7a577dae : SUCCESS in 5m 04s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/9cf8e830e4cc481b8fe4c19146cf8b96 : SUCCESS in 4m 13s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/4c6437d790854afd9ddd56e410be1f78 : SUCCESS in 3m 48s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c979c61370847ae85daac83a1529027 : SUCCESS in 2m 54s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/b5d346d07e3243d6bae630a1ea13325c : SUCCESS in 3m 12s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/29dc1a81046c437faa4d32a5c8f8f3c0 : SUCCESS in 3m 23s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/e75f2cc1edbd464a972acc5b779e0232 : SUCCESS in 3m 11s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/12a38a70aa604c07b7ac5708ef6caf17 : SUCCESS in 3m 23s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/6e59b5d38ad44ac191c1c91a5d309822 : SUCCESS in 2m 42s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/5a4f8c944280430d88cbb56d7978cbfd : SUCCESS in 5m 24s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/64d7badb9e5f4197b0183a443ea8a6da : SUCCESS in 6m 20s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8e2223dfa4146fa97732584e0c8df0d : SUCCESS in 5m 19s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbb551fa27704d13b98ea63cc6991ec4 : SUCCESS in 5m 20s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/c546ed49276b40f68770a0a3288b030d : SUCCESS in 6m 57s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/13679e038dfa45deb947b7d8c8b4cbb6 : SUCCESS in 6m 04s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/d694e5ab0dc14fc28870e21bf4a6fcd8 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":3},{"id":"96184ef108afd7e18fcbce6efbb50f40b0b9d885","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 18:58:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"84d936fa436e804700e0398f478fcb0868f708e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 19:12:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1adceb2778814dd98471e4748620a70d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bc5e9b0308db4405913878e7960f4301 : SUCCESS in 1m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc90b89cd23b402dabf8ecc4152fd4cb : SUCCESS in 4m 19s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/1d2d3fbe00f34a38815f38b390270714 : SUCCESS in 4m 18s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/2c8c66e5ddd14da6b810dd353acc0edd : SUCCESS in 3m 22s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/49b0eef924604b4487a3ba3883addaa7 : SUCCESS in 3m 14s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/d04300f6c92440db9cf0566554670224 : SUCCESS in 3m 27s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/50c264fb9b66484089cb1f85edeac6ab : SUCCESS in 3m 45s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/182337f3960b462e91bc6d460c014de3 : SUCCESS in 3m 43s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/cf77761bbdfc402696797311488b7ede : SUCCESS in 3m 49s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/01c46f844b6f4369aba68951c27a1be4 : SUCCESS in 5m 53s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/8a23383399814a1fb4a6723b25436c7c : SUCCESS in 6m 52s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/507dbdf00ee7432793a2050bfa4c545c : SUCCESS in 5m 27s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41e21dec963e4f6d9a66463e3376e9d2 : SUCCESS in 5m 37s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/284682edcf2447a793dd335fd2dc1949 : SUCCESS in 7m 42s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/582fa85da0ea49fbbfa0dfbfee6f1f5c : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":3},{"id":"d0b1065aa22eb2a05d7c7b19ec277cf9546f042c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 19:12:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0334380d3606fa3f1569fb77b3e466dfeffec2f2","revisions":{"22f9c9679a515cd9dc6bc5734d397f81b87a331c":{"kind":"REWORK","_number":1,"created":"2023-11-21 11:52:05.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/41/901541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/41/901541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/901541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/901541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/901541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/41/901541/1"}}},"commit":{"parents":[{"commit":"e8613ffeb01f109083f6a75dd148d5a8d37c9564","subject":"Propose rally 3.4.0 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e8613ffeb01f109083f6a75dd148d5a8d37c9564"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-11-21 11:51:57.000000000","tz":60},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-11-21 11:51:57.000000000","tz":60},"subject":"Drop support for py36 and py37","message":"Drop support for py36 and py37\n\nChange-Id: Ic16ce1d9b479dd53e1e157699dd80594318e77fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/22f9c9679a515cd9dc6bc5734d397f81b87a331c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/22f9c9679a515cd9dc6bc5734d397f81b87a331c"}]},"branch":"refs/heads/master"},"570d51bde3c953e9327078e0049d71af8acf1314":{"kind":"REWORK","_number":2,"created":"2023-11-21 11:57:35.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/41/901541/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/41/901541/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/901541/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/901541/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/901541/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/41/901541/2"}}},"commit":{"parents":[{"commit":"e8613ffeb01f109083f6a75dd148d5a8d37c9564","subject":"Propose rally 3.4.0 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e8613ffeb01f109083f6a75dd148d5a8d37c9564"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-11-21 11:51:57.000000000","tz":60},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-11-21 11:56:05.000000000","tz":60},"subject":"Drop support for py36 and py37","message":"Drop support for py36 and py37\n\nChange-Id: Ic16ce1d9b479dd53e1e157699dd80594318e77fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/570d51bde3c953e9327078e0049d71af8acf1314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/570d51bde3c953e9327078e0049d71af8acf1314"}]},"branch":"refs/heads/master"},"0334380d3606fa3f1569fb77b3e466dfeffec2f2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-11-21 11:58:08.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/41/901541/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/41/901541/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/901541/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/901541/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/901541/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/41/901541/3"}}},"commit":{"parents":[{"commit":"e8613ffeb01f109083f6a75dd148d5a8d37c9564","subject":"Propose rally 3.4.0 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e8613ffeb01f109083f6a75dd148d5a8d37c9564"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-11-21 11:51:57.000000000","tz":60},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-11-21 11:57:40.000000000","tz":60},"subject":"Drop support for py36 and py37","message":"Drop support for py36 and py37\n\nDepends-On: I16e4c61c8b166d5e128907d79ab6f857f98bd82b\nChange-Id: Ic16ce1d9b479dd53e1e157699dd80594318e77fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0334380d3606fa3f1569fb77b3e466dfeffec2f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0334380d3606fa3f1569fb77b3e466dfeffec2f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
