)]}'
{"id":"openstack%2Frally~902158","triplet_id":"openstack%2Frally~master~I26bf6a5847d1852ebd3badbb976320a01c49aa86","project":"openstack/rally","branch":"master","topic":"improve-subunit","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2023-11-29 15:38:58.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"I26bf6a5847d1852ebd3badbb976320a01c49aa86","subject":"Fix the issue when memoryview obj gets to subunit parser","status":"MERGED","created":"2023-11-29 15:02:11.000000000","updated":"2023-11-29 15:51:38.000000000","submitted":"2023-11-29 15:51:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902158-improve-subunit","meta_rev_id":"54636b6b80ffa64a4d3ddf03a8aede023bf3ae7a","_number":902158,"virtual_id_number":902158,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-29 15:51:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2023-11-29 15:38:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2023-11-29 15:38:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-29 15:27:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b33ac10b95a0e159c0743c8e01e3abbc6d7b9ef9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-11-29 15:02:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97edc96ac2653f4727248d63949a268b4159ca2c","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-11-29 15:02:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"26d4767c15920387e314f676f21d6132cce7eb03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-11-29 15:13:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7530edb47dc48f411ffd9ad20d9ed9bb2458e6a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 15:27:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aedb5c4e1e5c47f687cd59d759a17c6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e18ab2ea7c4d0eba2f706d926bd706 : SUCCESS in 4m 21s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/025bed9c46ee4f918f4d808344c3d735 : SUCCESS in 5m 35s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/06cdc51e08644d9c9144e505a7fe7948 : SUCCESS in 3m 47s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/35c1ffb9709147618ee154da16cdf8d8 : SUCCESS in 2m 56s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/96a09b535ce748fab9f7159643351a85 : SUCCESS in 2m 58s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/7d296f0957ee484980deccc310978cb1 : SUCCESS in 3m 09s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/8722ba2c4ef746b8a9ad876ad7da1330 : SUCCESS in 2m 56s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/14b621e62b6747cfa9ec38cf094ae95c : SUCCESS in 3m 26s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/fe6f51b636c84105b1fe983e71189f68 : SUCCESS in 2m 50s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/a66ac3371fef47c2988dc7e7db9e5e85 : SUCCESS in 5m 52s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ba123d9093344008b44c7b996f95e86e : SUCCESS in 6m 20s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c32d40455234b3cb501179b916ff4df : SUCCESS in 5m 57s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbbfa012e5894decba62e6ad38115394 : SUCCESS in 5m 39s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/6ad1b2ff879b4d7085ec68cb21487eca : SUCCESS in 7m 02s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/b86a1e5963ca451e8261621bb06759e9 : SUCCESS in 6m 16s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/df6048da929a4071a2a8fa288990d632 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":2},{"id":"c03e1f5fd7ccb5907459700584e9746b1e274b5a","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-11-29 15:38:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3a6a7c257e2dec4d8ac13e8d9a13407eda647a3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 15:39:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b8c4423de37995f31ed713033073ad4de74b4b6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 15:51:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d0136394f72401fb2a3b65412404e3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124aa08f7c36434bb9731932aafd9aa2 : SUCCESS in 6m 05s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/41401136c0644092b53a6d0ac4f77a66 : SUCCESS in 5m 11s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/c45e150c22b840b4bac49521c0afecc2 : SUCCESS in 3m 43s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e210092313a48f3984ec9a2e4952656 : SUCCESS in 2m 36s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/d1e0914934bd42fc80328baafa8bcb09 : SUCCESS in 3m 24s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/9798d0f9398746cf9f3a6045e159dedf : SUCCESS in 3m 24s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/958f265078944ea09ca08ab22cc577ca : SUCCESS in 3m 15s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/e41dbae817024faf87d0a7d3823d17a2 : SUCCESS in 3m 14s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/0872384e098e4c4ab2dc7c7008917b55 : SUCCESS in 6m 00s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/da0a65cea76e448086c9fc89809237dc : SUCCESS in 6m 41s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec2500f65f33421791594e102ee6b242 : SUCCESS in 5m 34s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cc3296ddbba43999f4efd1d7da86440 : SUCCESS in 5m 56s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/a7006be8f9614e339aaae07f3a031bec : SUCCESS in 6m 34s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/66968c101504494da37ee5d0ef3d1893 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":2},{"id":"54636b6b80ffa64a4d3ddf03a8aede023bf3ae7a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 15:51:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"46afd01f5f79517e4bbc1d2ee559bfb7cbbefc06","revisions":{"6e44209a6b1d7a620672c25c0e3fd518c8b23695":{"kind":"REWORK","_number":1,"created":"2023-11-29 15:02:11.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/58/902158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/58/902158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/58/902158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/58/902158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/58/902158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/58/902158/1"}}},"commit":{"parents":[{"commit":"0334380d3606fa3f1569fb77b3e466dfeffec2f2","subject":"Drop support for py36 and py37","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0334380d3606fa3f1569fb77b3e466dfeffec2f2"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-11-29 14:56:48.000000000","tz":60},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-11-29 15:01:34.000000000","tz":60},"subject":"Cover memoryview obj at subunit stream parser","message":"Cover memoryview obj at subunit stream parser\n\nChange-Id: I26bf6a5847d1852ebd3badbb976320a01c49aa86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6e44209a6b1d7a620672c25c0e3fd518c8b23695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6e44209a6b1d7a620672c25c0e3fd518c8b23695"}]},"branch":"refs/heads/master"},"46afd01f5f79517e4bbc1d2ee559bfb7cbbefc06":{"kind":"REWORK","_number":2,"created":"2023-11-29 15:13:36.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/58/902158/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/58/902158/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/58/902158/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/58/902158/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/58/902158/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/58/902158/2"}}},"commit":{"parents":[{"commit":"0334380d3606fa3f1569fb77b3e466dfeffec2f2","subject":"Drop support for py36 and py37","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0334380d3606fa3f1569fb77b3e466dfeffec2f2"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-11-29 14:56:48.000000000","tz":60},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-11-29 15:12:54.000000000","tz":60},"subject":"Fix the issue when memoryview obj gets to subunit parser","message":"Fix the issue when memoryview obj gets to subunit parser\n\nChange-Id: I26bf6a5847d1852ebd3badbb976320a01c49aa86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/46afd01f5f79517e4bbc1d2ee559bfb7cbbefc06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/46afd01f5f79517e4bbc1d2ee559bfb7cbbefc06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
