)]}'
{"id":"openstack%2Frally~902373","triplet_id":"openstack%2Frally~master~I0f9bf2a8d459bab22c3ebb926e1b9787b54c2eeb","project":"openstack/rally","branch":"master","topic":"py311","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2023-12-04 17:28:32.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"I0f9bf2a8d459bab22c3ebb926e1b9787b54c2eeb","subject":"[docker] use python3.11-slim as base image","status":"MERGED","created":"2023-12-01 11:45:50.000000000","updated":"2023-12-04 17:42:56.000000000","submitted":"2023-12-04 17:42:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902373-py311","meta_rev_id":"bed74258c232b0c5aa11bfbc1c632965dd9b1ebf","_number":902373,"virtual_id_number":902373,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-04 17:42:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2023-12-04 17:28:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2023-12-04 17:28:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-01 11:58:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a186a1334b961df0714a59687911be27bdd00555","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-12-01 11:45:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7da669b2783820150a5799b9629ed625b698771f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 11:58:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab142a85359544ea8b4d545c91437e46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/888fb0d954324b88abfc37bef7fe5e6f : SUCCESS in 5m 14s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/6e7894f79fb544b8899ed74b7a1c0ec4 : SUCCESS in 3m 41s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/71c7a8ba7b5146a79293981a98d759eb : SUCCESS in 3m 35s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/feb73691c8c64f96908fe6cbedb8bdea : SUCCESS in 2m 56s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/2e365201b4df4f7baf790dc96e10ede2 : SUCCESS in 3m 54s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/76b20159fedf4eb585432022cfccdd2b : SUCCESS in 3m 05s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/ec789927e801472e8681080225a86ab8 : SUCCESS in 4m 17s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/695cba93064d447fa531c4307b995aaa : SUCCESS in 3m 31s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/1bd9ecfa8639426a8260339942058986 : SUCCESS in 3m 28s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ac08853288bb485f8eb3e268d6638b95 : SUCCESS in 5m 23s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/d745782ca4a441af8516ab4b7326bdf0 : SUCCESS in 6m 17s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acfee38192e1467499f4dfbced7940d5 : SUCCESS in 5m 34s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6a07859256b4971b81646dd660ae80f : SUCCESS in 6m 10s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/e8c8e77dd72243d784bf6868d3d0a5da : SUCCESS in 7m 25s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/10fd14ed75754674b2b25e53a7b58a96 : SUCCESS in 6m 10s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/1c56d14e16fa40afb7a1d6f41cda15ba : FAILURE in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f0ebab9598a2276fc0fb7a810e7e11e3808b2c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-12-04 15:53:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"42f43ae1611444f137160dddbf9fb52c9f88ff26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 16:37:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73ff6651cd684a0298db17e5aa3327ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/018c6de2263c4d99bb8ed7f4d07056b8 : SUCCESS in 5m 04s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/fe766d9b5a124be1839897059fa963ea : SUCCESS in 4m 10s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f1f2408dab244093962ddcd8baf06733 : SUCCESS in 3m 46s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1952bb07f180435a9e083088278fabb4 : SUCCESS in 3m 08s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/8bc03c0c3f834fb598a1fb8255b31d40 : SUCCESS in 3m 52s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/61da5a02061c456098eeee9ac5690e7c : SUCCESS in 3m 34s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/b9c7fab0a47946c0b274b5fa803a60cc : SUCCESS in 3m 13s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/ef598fc1a5624656bcfaa193eb497290 : SUCCESS in 3m 06s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/cad52122b3834ec886150fd6e2512ef5 : SUCCESS in 4m 05s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/68202558313d4dfd8ab5de8affec6244 : SUCCESS in 5m 47s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/a83ed63c5543429bb6d63a410d071870 : SUCCESS in 8m 03s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa09cba4c906466a92a3e7b919936226 : SUCCESS in 5m 09s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f0f68f677404dbcb2c14b9e32bb1669 : SUCCESS in 5m 47s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/b5f905d5366043e08db75dcc1fa00ae0 : SUCCESS in 7m 06s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/61809beb41fc410fbdf93a47cda7a201 : SUCCESS in 6m 12s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/c82ceec2b26b465d87efa04bdbfffb2c : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":2},{"id":"96e97e385ded65d9c848fb052840a45700c068b1","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-12-04 17:28:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"54e5da1d5878691fc980829ddf51c23d1712b01a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 17:28:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"106fe84e4bd6e0d95f180a6ba1350f7fab33aeec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 17:42:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/376e4c5649f84e1f93adce8073f64b99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a73fa6ba6c4294a0be826705277383 : SUCCESS in 5m 11s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/88f62859e3fa40cb91cfbc983e576f93 : SUCCESS in 4m 13s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/87ef925268314890bef3e81c9ac729c9 : SUCCESS in 4m 40s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1312f48400ca40e7b2100ccedcd1e121 : SUCCESS in 3m 17s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/51629ba251da4bcab8db37fd6be9f200 : SUCCESS in 3m 11s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/e3b523f8b7a449ffbc7ba90034daf6a9 : SUCCESS in 3m 08s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/a6c7619ee1d0478fa5733afe38391d2b : SUCCESS in 3m 09s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/b7556509816f4765aa842ae47e117da4 : SUCCESS in 3m 56s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/bdd6fa9314c34760a9c708df3070d87d : SUCCESS in 6m 09s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/6dda8b0f58474b7aaed448213220aeaa : SUCCESS in 7m 29s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/958fec4f76f34524bb05ae26dda78fba : SUCCESS in 5m 43s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61707c5dcc3946548b4c73d13c4350b9 : SUCCESS in 5m 33s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/bdc3b55c4bec4a83b8d5d454602952a2 : SUCCESS in 7m 00s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/eca2eff847f74bf28e0a23f81e14a678 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":2},{"id":"bed74258c232b0c5aa11bfbc1c632965dd9b1ebf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 17:42:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2d7f51d2d761038a0eab4fd834c057432eb29117","revisions":{"2afe11771838b0e38733b0f16e44d27d54b11787":{"kind":"REWORK","_number":1,"created":"2023-12-01 11:45:50.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/73/902373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/73/902373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/73/902373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/73/902373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/73/902373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/73/902373/1"}}},"commit":{"parents":[{"commit":"46afd01f5f79517e4bbc1d2ee559bfb7cbbefc06","subject":"Fix the issue when memoryview obj gets to subunit parser","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/46afd01f5f79517e4bbc1d2ee559bfb7cbbefc06"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-12-01 11:45:42.000000000","tz":60},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-12-01 11:45:42.000000000","tz":60},"subject":"[docker] use python3.11-slim as base image","message":"[docker] use python3.11-slim as base image\n\nChange-Id: I0f9bf2a8d459bab22c3ebb926e1b9787b54c2eeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2afe11771838b0e38733b0f16e44d27d54b11787"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2afe11771838b0e38733b0f16e44d27d54b11787"}]},"branch":"refs/heads/master"},"2d7f51d2d761038a0eab4fd834c057432eb29117":{"kind":"REWORK","_number":2,"created":"2023-12-04 15:53:27.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/73/902373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/73/902373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/73/902373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/73/902373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/73/902373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/73/902373/2"}}},"commit":{"parents":[{"commit":"46afd01f5f79517e4bbc1d2ee559bfb7cbbefc06","subject":"Fix the issue when memoryview obj gets to subunit parser","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/46afd01f5f79517e4bbc1d2ee559bfb7cbbefc06"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-12-01 11:45:42.000000000","tz":60},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-12-04 15:53:15.000000000","tz":60},"subject":"[docker] use python3.11-slim as base image","message":"[docker] use python3.11-slim as base image\n\nChange-Id: I0f9bf2a8d459bab22c3ebb926e1b9787b54c2eeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2d7f51d2d761038a0eab4fd834c057432eb29117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2d7f51d2d761038a0eab4fd834c057432eb29117"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
