)]}'
{"id":"openstack%2Frally~912720","triplet_id":"openstack%2Frally~master~I0d4479537668c7b2980f64e6771413d9b4e82c83","project":"openstack/rally","branch":"master","attention_set":{},"removed_from_attention_set":{"36609":{"account":{"_account_id":36609,"name":"Youngjun Yoo","display_name":"Youngjun","email":"yj.yoo@okestro.com","username":"Youngjun"},"last_update":"2025-08-20 08:23:54.000000000","reason":"Change was submitted"},"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2025-08-20 07:59:36.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"I0d4479537668c7b2980f64e6771413d9b4e82c83","subject":"refectoring: refectoring docstring","status":"MERGED","created":"2024-03-13 08:42:59.000000000","updated":"2025-08-20 08:23:54.000000000","submitted":"2025-08-20 08:23:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912720","meta_rev_id":"ae49ce9661a9c4fcd362073f3ffa5fc7231d8961","_number":912720,"virtual_id_number":912720,"owner":{"_account_id":36609,"name":"Youngjun Yoo","display_name":"Youngjun","email":"yj.yoo@okestro.com","username":"Youngjun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-20 08:23:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2025-08-20 07:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2025-08-20 07:59:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-13 09:21:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-29 17:02:56.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"0042f427cd444872f363dd04b24e1e4c926d86ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36609,"name":"Youngjun Yoo","display_name":"Youngjun","email":"yj.yoo@okestro.com","username":"Youngjun"},"date":"2024-03-13 08:42:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a314f992e667d3d80906fbad9b48a90243303ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 09:21:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a2d4cbd2f3b4fd3ae97d99f1b8d55e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c7f2576d80b4f5098ff7585388fb7cb : SUCCESS in 5m 01s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e4a3204e98c94b778883e4419d48b747 : SUCCESS in 4m 43s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/5ed26899cb06410e8f20b5a1b34b2dc2 : SUCCESS in 3m 28s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c68dc08fc294b3d8b8d85a721ea34c8 : SUCCESS in 2m 37s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/a9c3a25954474d0e98f1a634b182b810 : SUCCESS in 2m 57s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/c930e4a7348e4c2d9cb84b2e903704a1 : SUCCESS in 2m 53s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/6a71b22a68cd4cb08b54e824fb42ff43 : SUCCESS in 3m 10s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/0d9a132081724419a4d0462e7c2ebab7 : SUCCESS in 2m 55s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/b55d92b6f4a44f49941155f988d402cf : SUCCESS in 2m 40s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/35f26f591266417aba9e52178d8a39cb : FAILURE in 5m 42s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/0cb2fab1202d4013a153fc8ad6262880 : SUCCESS in 5m 52s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe79089c90cb49e9aa11ebfc1ad63d2b : SUCCESS in 6m 00s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef1406bc19bc489781c06d3c44457ace : SUCCESS in 5m 30s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/331f90a44b054bd8b77d2c5ef88f1acc : SUCCESS in 6m 36s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/de2dd5b372794325a7fb412f2884f572 : SUCCESS in 6m 21s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/66fb6ec22d9d499dbd7f001802b776ed : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4a71aea301e727d7f009ace1bdec6d3b43333cb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36609,"name":"Youngjun Yoo","display_name":"Youngjun","email":"yj.yoo@okestro.com","username":"Youngjun"},"real_author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2024-04-29 17:02:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_36609\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":36609,"name":"Youngjun Yoo","display_name":"Youngjun","email":"yj.yoo@okestro.com","username":"Youngjun"}],"_revision_number":2},{"id":"a4bffc83d5e15b2cf969b99e9f383a0a0d59f677","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2024-04-29 17:02:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cc9c0e9e04a1ce3cd55449438db8c0bf5788fef9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 17:24:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02b521e6482e495db85ddfe4488e8cc9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d9bb59503ef45d198962cf18959ea69 : SUCCESS in 4m 18s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/770569c8862d497cbaa5cd41eff9b883 : SUCCESS in 4m 08s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/e2190cc5b4e04fa0a13955222cdf93dd : SUCCESS in 3m 52s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/be673607873f4fcd9f264a1f4e31516b : SUCCESS in 2m 40s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/a34c9a27db0b4eb58b8cb80318718766 : FAILURE in 3m 05s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/5267743f7ba24570ba9acfc11cb4bb29 : FAILURE in 2m 52s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/9571af30ff684486812a5f7b64c7e47d : FAILURE in 2m 42s\n- rally-tox-py311-sqlalchemy14 https://zuul.opendev.org/t/openstack/build/82ea3b81800e4d1dbacd13ce95d58c24 : FAILURE in 2m 57s\n- rally-tox-py311-sqlalchemy2 https://zuul.opendev.org/t/openstack/build/c70d3add12d84967bcabcbe8cb201614 : FAILURE in 2m 46s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/e55b8776ff004b279876fc126a40a3ae : FAILURE in 2m 42s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/f78724bb866f442d82ba610d64040d71 : FAILURE in 2m 41s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/588b51f87ef4419ba03b34f30c7e8969 : SUCCESS in 6m 11s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96db852e6e3049eba190795ba1dbfbf9 : SUCCESS in 5m 50s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45b5d7e828d74219a958f4365a044704 : SUCCESS in 5m 35s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/4277c1d721cd430d9bdc50a9978609ce : SUCCESS in 6m 20s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/9de333816a244827bc09b1feccaca0c2 : SUCCESS in 6m 05s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/e3edf14e6ba649de81bc3c16bf70263c : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e812b1af5724dee565a6a2ec67ca1b9516e2ba31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36609,"name":"Youngjun Yoo","display_name":"Youngjun","email":"yj.yoo@okestro.com","username":"Youngjun"},"real_author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-08-20 07:59:28.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_36609\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[{"_account_id":36609,"name":"Youngjun Yoo","display_name":"Youngjun","email":"yj.yoo@okestro.com","username":"Youngjun"}],"_revision_number":3},{"id":"d0a5cd9f3c6f26f7dbbb4565c05ee4be56f44300","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-08-20 07:59:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2d39701e726683242737f3b599464fbfb983380c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 08:11:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bccbd2436e4444a1b634f62b33a9a9ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f50f532da144da894444718a3e31380 : SUCCESS in 4m 19s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/7ae259c6b22c4ad58472f38e4c51f31c : SUCCESS in 2m 51s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/162e067603ad48faaf3a9fa78b0fbe2b : SUCCESS in 4m 52s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/71f46626c86d41c38ba403222d585add : SUCCESS in 2m 30s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8f2815bc8fe4bf3a1db6102cffe3ac4 : SUCCESS in 2m 41s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/82b1790321f14d7680f5fd1ceafd8283 : SUCCESS in 2m 29s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/7a461ea15a4a4bc5bdf7d5aae6f36bc8 : SUCCESS in 5m 31s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/8758730a732b48039a2d1b051ad93cf7 : SUCCESS in 3m 24s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/a87859147ecc48c4b6af9477c00ecadc : SUCCESS in 4m 07s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/6f9d764906564ca4a4453d43d250a5c7 : SUCCESS in 3m 06s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/cbfb46d1581846a59e5d74bd569d0429 : SUCCESS in 7m 36s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/def71a404c7547a58c085b158ccd144e : SUCCESS in 6m 31s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6466b93aea2469b8d680173bb4059bf : SUCCESS in 4m 50s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/875e75eaca7543eda0ac5557dd14c4dd : SUCCESS in 5m 05s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/71ca370424484cf3bc29d2c998053b94 : SUCCESS in 7m 17s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/e83e9bb1151f44c2bc6a5745a47cdc2a : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":3},{"id":"77a8cf0102f6167be66a5873ba7c5706d624c0a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 08:11:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4fe32fc3be513ec43870594063d0bfb1bd560059","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 08:23:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/066f213c4bb84fb2ba249d457349158b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b11f0ca484cc41f399cec8d5484f35b0 : SUCCESS in 4m 00s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/1dcab11e57d5483f8dfa70df74ffd7ab : SUCCESS in 4m 05s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/25cbda87f7d944d39f343d4639e32f9f : SUCCESS in 3m 13s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/c15f9523af654f3793b064556344e864 : SUCCESS in 3m 06s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/489e49c0d894469b8cc0927c7099428f : SUCCESS in 2m 57s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/2ca0c56287f647e0a67d4692c7467500 : SUCCESS in 3m 47s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/a9606c8f585f4703b2d3bcbef8b19fd3 : SUCCESS in 3m 11s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/bbebf64d6ec1496ea3c176a8a4028321 : SUCCESS in 3m 26s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/75719a7eefa242689fcbebbf897edf9f : SUCCESS in 4m 33s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/2a212b3d59364244a9d38cbb6bb8b2ae : SUCCESS in 3m 07s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/33cb7519e88141b581cd0e2480f4beea : SUCCESS in 5m 19s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/3839b64c4fb64dab91d63cf030cd1b43 : SUCCESS in 7m 02s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea22bf20ad8c4f8bb82ecb91c8028e5a : SUCCESS in 5m 19s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ad2d11f22944a128b71dbd85d4bbdad : SUCCESS in 8m 34s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/de9a5ef523af4b249015c26f7d9bf43d : SUCCESS in 6m 00s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/550f3e56883c4dcb9152086ddd4b5b45 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":3},{"id":"ae49ce9661a9c4fcd362073f3ffa5fc7231d8961","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 08:23:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d87fb8ec3072c4d39ac9ff2f842334a5cd699c66","revisions":{"7fb428fd673d28a62afff2ddb0f88752d217cc77":{"kind":"REWORK","_number":1,"created":"2024-03-13 08:42:59.000000000","uploader":{"_account_id":36609,"name":"Youngjun Yoo","display_name":"Youngjun","email":"yj.yoo@okestro.com","username":"Youngjun"},"ref":"refs/changes/20/912720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/20/912720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/912720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/912720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/912720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/20/912720/1"}}},"commit":{"parents":[{"commit":"7ad868bb9bc7834fa573f3612b8f3f22e2bb9719","subject":"Update u-c file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7ad868bb9bc7834fa573f3612b8f3f22e2bb9719"}]}],"author":{"name":"Youngjun","email":"yj.yoo@okestro.com","date":"2024-03-13 08:41:56.000000000","tz":540},"committer":{"name":"Youngjun","email":"yj.yoo@okestro.com","date":"2024-03-13 08:42:56.000000000","tz":540},"subject":"refectoring: refectoring docstring","message":"refectoring: refectoring docstring\n\nChange-Id: I0d4479537668c7b2980f64e6771413d9b4e82c83\nSigned-off-by: Youngjun \u003cyj.yoo@okestro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7fb428fd673d28a62afff2ddb0f88752d217cc77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7fb428fd673d28a62afff2ddb0f88752d217cc77"}]},"branch":"refs/heads/master"},"278a9ea599aaf9250fd74b5f0a5e4390187374f0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-29 17:02:51.000000000","uploader":{"_account_id":36609,"name":"Youngjun Yoo","display_name":"Youngjun","email":"yj.yoo@okestro.com","username":"Youngjun"},"real_uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/20/912720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/20/912720/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/912720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/912720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/912720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/20/912720/2"}}},"commit":{"parents":[{"commit":"e5d9fc8e63cc0727831633cf5b89e5114270c8c3","subject":"Add support for SQLAlchemy 2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e5d9fc8e63cc0727831633cf5b89e5114270c8c3"}]}],"author":{"name":"Youngjun","email":"yj.yoo@okestro.com","date":"2024-03-13 08:41:56.000000000","tz":540},"committer":{"name":"Youngjun Yoo","email":"yj.yoo@okestro.com","date":"2024-04-29 17:02:51.000000000","tz":0},"subject":"refectoring: refectoring docstring","message":"refectoring: refectoring docstring\n\nChange-Id: I0d4479537668c7b2980f64e6771413d9b4e82c83\nSigned-off-by: Youngjun \u003cyj.yoo@okestro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/278a9ea599aaf9250fd74b5f0a5e4390187374f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/278a9ea599aaf9250fd74b5f0a5e4390187374f0"}]},"branch":"refs/heads/master","description":"Rebase"},"d87fb8ec3072c4d39ac9ff2f842334a5cd699c66":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-08-20 07:59:28.000000000","uploader":{"_account_id":36609,"name":"Youngjun Yoo","display_name":"Youngjun","email":"yj.yoo@okestro.com","username":"Youngjun"},"real_uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/20/912720/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/20/912720/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/912720/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/912720/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/912720/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/20/912720/3"}}},"commit":{"parents":[{"commit":"6565717213fa7b87f383872d5d639b4d87c5faf2","subject":"Port all samples to task v2 format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6565717213fa7b87f383872d5d639b4d87c5faf2"}]}],"author":{"name":"Youngjun","email":"yj.yoo@okestro.com","date":"2024-03-13 08:41:56.000000000","tz":540},"committer":{"name":"Youngjun Yoo","email":"yj.yoo@okestro.com","date":"2025-08-20 07:59:28.000000000","tz":0},"subject":"refectoring: refectoring docstring","message":"refectoring: refectoring docstring\n\nChange-Id: I0d4479537668c7b2980f64e6771413d9b4e82c83\nSigned-off-by: Youngjun \u003cyj.yoo@okestro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d87fb8ec3072c4d39ac9ff2f842334a5cd699c66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d87fb8ec3072c4d39ac9ff2f842334a5cd699c66"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
