)]}'
{"id":"openstack%2Frally~913186","triplet_id":"openstack%2Frally~master~Ib32974db5cfebf964c9c0e450a63174b5d999fb3","project":"openstack/rally","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-03-27 21:54:16.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2024-03-27 21:54:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib32974db5cfebf964c9c0e450a63174b5d999fb3","subject":"Fix ModuleNotFoundError: No module named \u0027fixtures\u0027","status":"ABANDONED","created":"2024-03-14 13:36:07.000000000","updated":"2024-04-29 20:23:44.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"77cccdbfa0ef90f5cc68ccc248cc38f6146b63df","_number":913186,"virtual_id_number":913186,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-14 14:19:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"disliked":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":-1,"date":"2024-03-27 20:24:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":1,"date":"2024-03-25 08:13:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-14 14:19:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-25 08:13:32.000000000","updated_by":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"reviewer":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"state":"REVIEWER"},{"updated":"2024-03-27 20:24:42.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"655d315f2b9689cb7ab1746136c99021006d2b46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-14 13:36:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7b6a60f4695108e8b1b7dc2443d8ecc2ec20ab3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 14:19:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea235f6ccffb4e6aa98bc7ae273e6dd9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb22c742f8c465881b40dfc556c9afe : SUCCESS in 4m 07s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/19e97f631b964f6f89b3bcacf99e075a : SUCCESS in 4m 09s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/0dde2954745f4d66bf2bde07a69008b9 : SUCCESS in 3m 57s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa5a104515cd4854bbd471cd910c3bc0 : SUCCESS in 2m 47s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/a20c1128f2114984b6f46d8d726d88fa : SUCCESS in 2m 48s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/2e68d9c3d7e04d51aa8d92729c2ae8b8 : SUCCESS in 4m 09s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/2ee2691c51904d8ca864c8f74a592030 : SUCCESS in 3m 04s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/5de10e7fde184e7cb7124bc363127b94 : SUCCESS in 3m 01s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/b5fe3939fe9b4cf49b00a7acb59e18d9 : SUCCESS in 2m 40s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/879561ed76724c3a9cd28e9bedc16ee4 : FAILURE in 7m 35s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ba8b79cad1ed477eac6cbbdcc988f916 : SUCCESS in 6m 03s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bb145c6e0de4847b5788ac4f52a3570 : SUCCESS in 5m 27s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/722683594877469d8cee57f91fe25252 : SUCCESS in 5m 37s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/a57cb868a37f4c7fbd748b3f13a8b38d : SUCCESS in 8m 05s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/73f865090ee140528162cc9550027589 : SUCCESS in 6m 27s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/651f073e2f3142369a5fae79df7a1b7d : SUCCESS in 8m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e0a09245798a93e6a4c820494d5bffa6f77833cd","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-25 08:13:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f2502ca3086fadfcf10308f77a2b48add9d584f9","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2024-03-27 20:24:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9134a07ee996d361ca10950b0fcd9c0f270bebb3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-27 21:26:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc560d9c0e3feb49f681112772c2da7ec7d63a22","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2024-03-27 21:42:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c20ec0b70f925b458c7e642521d53a5900f25ca","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-27 21:54:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bed128e2567cc4d48532afa12fa0c6740dfaf0c4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-27 21:54:41.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/rally/+/906038 fixes the issue","accounts_in_message":[],"_revision_number":1},{"id":"26a30183bcd10529b8056cf76ce074d1d8919eb0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-27 21:55:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77cccdbfa0ef90f5cc68ccc248cc38f6146b63df","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2024-04-29 20:23:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b8d5c28c78d886147ead9303421ed12df520407","revisions":{"9b8d5c28c78d886147ead9303421ed12df520407":{"kind":"REWORK","_number":1,"created":"2024-03-14 13:36:07.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/86/913186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/86/913186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/86/913186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/86/913186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/86/913186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/86/913186/1"}}},"commit":{"parents":[{"commit":"7ad868bb9bc7834fa573f3612b8f3f22e2bb9719","subject":"Update u-c file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7ad868bb9bc7834fa573f3612b8f3f22e2bb9719"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-03-14 13:35:01.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-03-14 13:35:47.000000000","tz":180},"subject":"Fix ModuleNotFoundError: No module named \u0027fixtures\u0027","message":"Fix ModuleNotFoundError: No module named \u0027fixtures\u0027\n\nCloses-Bug: #2057930\nChange-Id: Ib32974db5cfebf964c9c0e450a63174b5d999fb3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9b8d5c28c78d886147ead9303421ed12df520407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9b8d5c28c78d886147ead9303421ed12df520407"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
