)]}'
{"id":"openstack%2Frally~916576","triplet_id":"openstack%2Frally~master~I45600b817cada2266341afa0ac9dbfec6ce7e62e","project":"openstack/rally","branch":"master","topic":"sqlalchemy-upgrade","hashtags":[],"change_id":"I45600b817cada2266341afa0ac9dbfec6ce7e62e","subject":"Revert \"Remove usage of LegacyEngineFacade\"","status":"MERGED","created":"2024-04-20 14:25:20.000000000","updated":"2024-04-20 15:46:24.000000000","submitted":"2024-04-20 15:46:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916576-sqlalchemy-upgrade","meta_rev_id":"06ec2d61ef0f2b634a13b23b28d8e779b01df42c","_number":916576,"virtual_id_number":916576,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-20 15:46:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2024-04-20 15:05:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2024-04-20 15:05:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-20 14:50:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7d889797df414fc55b234d3f77ca4eb03367ba5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2024-04-20 14:25:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"135dd364fc6cf95c008f403e4ec5a9c792d3d730","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-20 14:50:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07faf30fc00145c9a98f2cc1b5dfca0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d7b14bc8ba4aa6baea8e05c97b848a : SUCCESS in 4m 20s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/abab87ba354c46268a6244080c5584e1 : SUCCESS in 4m 14s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/d9cc88eb8b2544c4b270201aac7be088 : SUCCESS in 3m 50s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc936cfb70a445a5b42e8ab6ae38a75f : SUCCESS in 2m 33s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/55bfe27b9cbf475f8f5d4f55e2db4497 : SUCCESS in 3m 20s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/54e5abe2607942e9956ac603e92a1f6b : SUCCESS in 2m 54s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/df646846a9e54a51bba301871b1ae6b1 : SUCCESS in 3m 06s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/0ff1b0922b424407ad2ea368fd6c1fec : SUCCESS in 2m 59s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/f486cfa9a09a45c89b3a9c29a37291a0 : SUCCESS in 2m 48s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/84b0d7baf976404caaa41d80d2e2272a : SUCCESS in 5m 36s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/8e1a39dde17a4a7793ba4a36c64402e9 : SUCCESS in 5m 59s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d96badb4c6145b3b1d2e7b843b29475 : SUCCESS in 5m 45s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52fca35be5d34a64960ecc3a85f1ea65 : SUCCESS in 5m 28s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/2b3519954b94483ca80e7e87470e5f5c : SUCCESS in 6m 34s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/d931fa27ede24c4bafce01cbc8730d54 : SUCCESS in 7m 06s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/591f2586aa70492ba643238f0c29d3ea : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0949cec17ed2d1ba0fb621a6032f7f1d6dd2392c","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2024-04-20 15:05:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"41463901d7e5128132cfda2b241c61cda2cd865a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-20 15:05:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14471f34f3e06959b3ce481c44ef112012741deb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-20 15:46:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a15338add8394783b1f253e362df440c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e38af25b3c9b4bf38b3a6fb6fef2d0cc : SUCCESS in 4m 40s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/9adc2a5fa2934977b4bf8a56652de645 : SUCCESS in 3m 50s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/e4881bdc24e0433ca44e9ff7f0c5ace1 : SUCCESS in 3m 50s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e548cb08994ee19abef446ab0ded43 : SUCCESS in 3m 47s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/abcf712a5720444cb78ff417559d2f21 : SUCCESS in 3m 36s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/ae9d7eda99e34ee9bce6e2148835cad8 : SUCCESS in 2m 54s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/e8305ed007c24d338621926ad2eb1f0d : SUCCESS in 3m 07s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/9916e9efae1d443ebcf3505436ae51a6 : SUCCESS in 2m 49s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/6f44ab66cb154c7abaede33eb52687cc : SUCCESS in 5m 43s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/6edeef7750b447e7b7c62d05e1e90e22 : SUCCESS in 6m 13s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1c0fb5b6cc643c5832f78bc14143a2d : SUCCESS in 8m 08s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4705757535fc4c8191e9b1dd4e0fc589 : SUCCESS in 6m 09s\n- rally-install-centos-8s https://zuul.opendev.org/t/openstack/build/fcc6d83e7c564215b08a113d520ee20c : SUCCESS in 6m 18s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/8fd39dd72a954a3c881b614dc674b69d : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"06ec2d61ef0f2b634a13b23b28d8e779b01df42c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-20 15:46:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96345af0cb157dc336a4a2e874d2bc803e0e9c3b","revisions":{"96345af0cb157dc336a4a2e874d2bc803e0e9c3b":{"kind":"REWORK","_number":1,"created":"2024-04-20 14:25:20.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/76/916576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/76/916576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/76/916576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/76/916576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/76/916576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/76/916576/1"}}},"commit":{"parents":[{"commit":"7ad868bb9bc7834fa573f3612b8f3f22e2bb9719","subject":"Update u-c file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7ad868bb9bc7834fa573f3612b8f3f22e2bb9719"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2024-04-20 14:18:36.000000000","tz":-240},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2024-04-20 14:18:49.000000000","tz":-240},"subject":"Revert \"Remove usage of LegacyEngineFacade\"","message":"Revert \"Remove usage of LegacyEngineFacade\"\n\nTemporary reverts commit 01018c1b1ea5dbe7fd360f62de9fbf6f239a34ed.\nThis change will be added back as a part of bigger commit\n\nChange-Id: I45600b817cada2266341afa0ac9dbfec6ce7e62e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/96345af0cb157dc336a4a2e874d2bc803e0e9c3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/96345af0cb157dc336a4a2e874d2bc803e0e9c3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
